)]}'
{"id":"openstack%2Fmistral~959376","triplet_id":"openstack%2Fmistral~master~I8ab94e24163fd1ef7bb0d26f324fde21a14cec78","project":"openstack/mistral","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-04 17:36:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I8ab94e24163fd1ef7bb0d26f324fde21a14cec78","subject":"Revert \"Disable python unit tests temporarily\"","status":"MERGED","created":"2025-09-03 19:13:10.000000000","updated":"2025-09-04 19:07:54.000000000","submitted":"2025-09-04 19:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959376-eventlet-removal","meta_rev_id":"b6da65abbecebf9267872db4421f60c87e08b8ae","_number":959376,"virtual_id_number":959376,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 19:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 08:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 08:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 20:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"931961a488d49549e5d8dbc70cf1ea1575db4960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 19:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dfd76d428b619fbdd8b2aeeb2404c465cfe6352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:07:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13f254e985c94b07b973985a98e1d6f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504f5d43790744bf9143cfb2bee2a8bb : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/275e294e56754e70849206f1022c1fea : TIMED_OUT in 40m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14742709f615446d88917539342fe4ba : TIMED_OUT in 40m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4457ead7e93f462cbb2ee0aedfad3a61 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf60cabdc69486491f7c69868308055 : SUCCESS in 8m 01s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/441afff64622477bb4d053396162d661 : FAILURE in 42m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71b270f4794a451ab76fd84f5f1f3076 : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c77ca8038a5a50cafffda8bb8f79713117ec4b73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 20:12:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"02fa4480cc2b73fbbfd296893ee8de24c6bdf1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:43:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20fb33fd2a442509a7c69c6a664d6a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adaee17fafa54463b73effd930f52fde : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d18be179d666457eb188a03d38f95a36 : SUCCESS in 9m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45eefbec66d24a6fa0bf27749a66f712 : SUCCESS in 12m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc0e8b0c86b84bbea5d10090b9a316ff : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98d35346df44399822d00e56cd7e4e5 : SUCCESS in 9m 10s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ef4732bc9a5e48d589781a73a53e000f : SUCCESS in 21m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/083245019dc24dfda494d71969278c09 : SUCCESS in 25m 56s","accounts_in_message":[],"_revision_number":2},{"id":"152d1cad5f766a774bc5c32c2b2e60cda49a7eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-03 22:04:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bfa6d14f625b851aba184bd1b769f0bb896ba756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 22:56:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/481fec6646444e28b8b4628eaa6ed4df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32e8fc0aa694839934567b4afca7103 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/044b594e0bc64cbe8d3da69726e1641b : SUCCESS in 8m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40606d3369a14f88a10fcda2f0100e0b : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7de9c4c768b4f63a2ec7806e1936f49 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f71dfe01504388b8bdbb589c80b2ab : SUCCESS in 9m 41s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/225884d2e2c64764a062b39825909e76 : SUCCESS in 41m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70c993f7accf431db817ac435d6efef2 : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d8b24d95e028546b90e87d362958216056d3ca37","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 08:50:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e45a7471be556c81595ed1de9c2bd13e56fa9cd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:51:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"307e3f1c2bb7caadb42d0aa2c5cfd2e7b30a3351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 09:41:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f3d826d17ab48b4b3a2fe1a1e7c9208\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e89fb36170142e3b57ab61cdf397703 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e2c45d3276f4c79a3951fe782ecd284 : SUCCESS in 10m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4618bcd663524e389860ace42bceb961 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4596a92a1847c3a0ea8c79c9ae59cf : SUCCESS in 7m 41s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5d74ce514589418c950abf763fbb127e : SUCCESS in 16m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0ae336a1b2142a7930d68e46f3a965f : FAILURE in 38m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9674f925144a786bf7f412fd1893c9bbe215bef1","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 13:49:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eabb1198978a71a90ca27004e9693cdfaccbc1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:38:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddb5f107a93f460ea802a38e9ecbe283\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6670545ed88443f89c7da176fa7814f : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecde71a226044d5b9724eb83b43357ba : SUCCESS in 11m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/794e61dc1dff4ac2a793542bb6fb135d : SUCCESS in 10m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec49dcc0572743cdb6fad9fa5b4ea88a : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f68b812ae44b0b9073eea35cbd222c : SUCCESS in 8m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/104e0c19b8ab49b19c6efbfca1de92b6 : SUCCESS in 40m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72c048ab585d48efacf36b8c115c47f3 : FAILURE in 42m 01s","accounts_in_message":[],"_revision_number":3},{"id":"bf7eb1129fbfe3e2573e3f2df43e133ab5394db0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-04 17:36:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47b47cb620d08eb3a0a15dd49b2f220bdb6d76f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:21:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea4007178d14edb8c918158d1dd410f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc2722e09f74b5c928eee42d6f25402 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/315048079fb542fc9917fd0de1e1f503 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf9aa68bc6e2403cbda238f93131bbf3 : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df2afee3ed32447daebed1cf89121dd4 : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6643d7c96acc43848779e438aaaf555b : SUCCESS in 7m 39s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/22b6d9e5bede49f082dcba7dc51931e8 : SUCCESS in 19m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0be340d1613b41ebb77a54a4a111d423 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":3},{"id":"12513d3324c5004b9dc847c9525538543f47b31a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 18:22:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"437f9f1c668c716532f0d2c84b8f7cc06f07b682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:05:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0fa4a9d0104c4d8bb60c35a7058336\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd71116011e0465091f19db817a9574e : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84fb7c22113542558970b3243725210f : SUCCESS in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/965327d418784801a0cda058f30642ca : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c18a87288e42d58ef4e3090d72eed4 : SUCCESS in 5m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1dd1423ce2c5427098ab06ecc14603b3 : SUCCESS in 22m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f4e69cf0caf4813ada828393ad8f547 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e9121311e63650364c07486b033b027edfb1e566","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:05:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b6da65abbecebf9267872db4421f60c87e08b8ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 19:07:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9673d2fa526d4e2ca8b00d165121d051\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071a9497d74c4271bf5657d5bfe5b371 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ec6f9c0a79c5e33e200baadc6e849e507e897f5","revisions":{"172ffd49cc72fab95b79ae2c63cad17873429e29":{"kind":"REWORK","_number":1,"created":"2025-09-03 19:13:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/76/959376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/959376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/959376/1"}}},"commit":{"parents":[{"commit":"ae43e217151d3dd7b49b4cb616896fc0231f2cf6","subject":"Replace eventlet wsgi by cheroot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ae43e217151d3dd7b49b4cb616896fc0231f2cf6"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:08:28.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:12:52.000000000","tz":120},"subject":"Revert \"Disable python unit tests temporarily\"","message":"Revert \"Disable python unit tests temporarily\"\n\nThis reverts commit 073a220cfaaf387c08c67c042073d66f50b32a94.\n\nChange-Id: I8ab94e24163fd1ef7bb0d26f324fde21a14cec78\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/172ffd49cc72fab95b79ae2c63cad17873429e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/172ffd49cc72fab95b79ae2c63cad17873429e29"}]},"branch":"refs/heads/master"},"485dd1b40fdceabfac258fd8a03528cd848f3d37":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-03 20:12:20.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/76/959376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/959376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/959376/2"}}},"commit":{"parents":[{"commit":"3bbcc221e8a8c73de4269c712395b10c2dcbf633","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3bbcc221e8a8c73de4269c712395b10c2dcbf633"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:08:28.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 20:12:10.000000000","tz":120},"subject":"Revert \"Disable python unit tests temporarily\"","message":"Revert \"Disable python unit tests temporarily\"\n\nThis reverts commit 073a220cfaaf387c08c67c042073d66f50b32a94.\n\nChange-Id: I8ab94e24163fd1ef7bb0d26f324fde21a14cec78\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/485dd1b40fdceabfac258fd8a03528cd848f3d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/485dd1b40fdceabfac258fd8a03528cd848f3d37"}]},"branch":"refs/heads/master"},"2ec6f9c0a79c5e33e200baadc6e849e507e897f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-03 22:04:13.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/76/959376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/959376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/959376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/959376/3"}}},"commit":{"parents":[{"commit":"bce61d6b2cf7415028676eaf02cac2e519dcc784","subject":"Rework event engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bce61d6b2cf7415028676eaf02cac2e519dcc784"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 19:08:28.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-03 22:03:31.000000000","tz":120},"subject":"Revert \"Disable python unit tests temporarily\"","message":"Revert \"Disable python unit tests temporarily\"\n\nThis reverts commit 073a220cfaaf387c08c67c042073d66f50b32a94.\n\nChange-Id: I8ab94e24163fd1ef7bb0d26f324fde21a14cec78\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ec6f9c0a79c5e33e200baadc6e849e507e897f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ec6f9c0a79c5e33e200baadc6e849e507e897f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
