)]}'
{"id":"openstack%2Fmistral~962032","triplet_id":"openstack%2Fmistral~master~I51f4128dd023846a4406cd1b414a56d741035483","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I51f4128dd023846a4406cd1b414a56d741035483","subject":"Do not call setup_db from engine","status":"MERGED","created":"2025-09-22 20:55:37.000000000","updated":"2025-09-23 06:48:44.000000000","submitted":"2025-09-23 06:47:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962032","meta_rev_id":"c4445b0eed8d35577612568769722648a63c0508","_number":962032,"virtual_id_number":962032,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-23 06:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-23 06:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-23 06:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-22 21:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"571843a074abcbae92321475c89bb16baecfc50d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-22 20:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdb62803a79984c91a2f1034a56fa1b794f99d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-22 20:57:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07aa39df8a257dab9699b136d98958543b3aa9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 21:19:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c4b19f416f4bd0910b87f0a1105b61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322fdf22674b4310b9918281a4d332dc : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/475ab9f19ca74e1eac154214db290f79 : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/562e5e4955f744bb91be5b6afa433b9a : SUCCESS in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24219a6ed3094076bb7c41a400f513e9 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5f83e251a045efafbce1602fca9231 : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/065c658d5506451c86bbcc9388bdd7de : SUCCESS in 17m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a83decbea9e64301a5b4494b24663738 : SUCCESS in 18m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4136232238264f68a32eba4f46d79fe0 : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"84169933efda6b3079ab87e7b2a8051afbbd6eaa","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-23 06:00:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae6dd7bc90a1e7aab2432b495f982f21148cce95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 06:01:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25ba0a22ca682fb6a8bf999e8907e6e272e771b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 06:47:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9986bea59a2e4c5ebd596cc9ae0b19da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7837967d4846a194de3c1088237626 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76ec7347f62f401c939490832e18478f : SUCCESS in 11m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1572dc22793d4651b8c52f509dbc51d1 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c5e9191e574b3a97583643958cdd1f : SUCCESS in 8m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d74731d10a6e4a2eb36f307d614c3c95 : SUCCESS in 24m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f0a2cccb682407792b9b7c8b2351192 : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a3c3b83eebca737dd545e1099194171124990e59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 06:47:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c4445b0eed8d35577612568769722648a63c0508","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 06:48:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902c6d41167844d896837c5bfdcfa8d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ecc313efec6402ba551944babe46a46 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8be59adc3810e8a68c94b6c63998eef7fb769830","revisions":{"d5fd2225be1898c757cab2860672479340df4ead":{"kind":"REWORK","_number":1,"created":"2025-09-22 20:55:37.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/962032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/962032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/962032/1"}}},"commit":{"parents":[{"commit":"d2994f8e84e6bfa55306a46df23abed720152156","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2994f8e84e6bfa55306a46df23abed720152156"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-22 20:32:09.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-22 20:53:38.000000000","tz":120},"subject":"Do not call setup_db from engine","message":"Do not call setup_db from engine\n\nCalling setup_db is creating the db structure if missing (because it\u0027s\ncalling models.Workbook.metadata.create_all after).\n\nThis is not what we want since we want the operator to manage the\ncreation of the db using mistral-db-manage like any other regular\nopenstack service.\n\nI suspect that this call was mostly useful for tests.\nThis is a real issue in production because it create the DB structure\nwithout using the correct alembic migration path (and thus,\nalembic_version table is missing).\n\nChange-Id: I51f4128dd023846a4406cd1b414a56d741035483\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5fd2225be1898c757cab2860672479340df4ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5fd2225be1898c757cab2860672479340df4ead"}]},"branch":"refs/heads/master"},"8be59adc3810e8a68c94b6c63998eef7fb769830":{"kind":"REWORK","_number":2,"created":"2025-09-22 20:57:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/32/962032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/962032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/962032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/962032/2"}}},"commit":{"parents":[{"commit":"d2994f8e84e6bfa55306a46df23abed720152156","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d2994f8e84e6bfa55306a46df23abed720152156"}]}],"author":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-22 20:32:09.000000000","tz":120},"committer":{"name":"Arnaud M","email":"arnaud.morin@gmail.com","date":"2025-09-22 20:57:37.000000000","tz":120},"subject":"Do not call setup_db from engine","message":"Do not call setup_db from engine\n\nCalling setup_db is creating the db structure if missing (because it\u0027s\ncalling models.Workbook.metadata.create_all after).\n\nThis is not what we want since we want the operator to manage the\ncreation of the db using mistral-db-manage like any other regular\nopenstack service.\n\nI suspect that this call was mostly useful for tests.\nThis is a real issue in production because it create the DB structure\nwithout using the correct alembic migration path (and thus,\nalembic_version table is missing).\n\nChange-Id: I51f4128dd023846a4406cd1b414a56d741035483\nSigned-off-by: Arnaud M \u003carnaud.morin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8be59adc3810e8a68c94b6c63998eef7fb769830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8be59adc3810e8a68c94b6c63998eef7fb769830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
