)]}'
{"id":"openstack%2Fmistral~963814","triplet_id":"openstack%2Fmistral~master~If71f5663c26457a5cf6b9eb88f15355e664b520e","project":"openstack/mistral","branch":"master","topic":"return_result_on_error","attention_set":{},"removed_from_attention_set":{"21970":{"account":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"last_update":"2025-10-31 21:33:22.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2025-10-31 21:33:22.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-10-31 21:33:22.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-10-31 20:46:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"If71f5663c26457a5cf6b9eb88f15355e664b520e","subject":"SSHAction: Add return_result_on_error\u003dTrue/False","status":"MERGED","created":"2025-10-12 21:35:25.000000000","updated":"2025-10-31 21:34:27.000000000","submitted":"2025-10-31 21:33:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963814-return_result_on_error","meta_rev_id":"7a31065266d8adf6b4c31402e551ade5227ba90f","_number":963814,"virtual_id_number":963814,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 21:33:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 20:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-31 20:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-12 22:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-15 12:25:41.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-10-20 14:03:55.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2025-10-20 14:03:55.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2025-10-31 20:46:04.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"57771c0aa5391dcfae60d7738eea00c87968ff72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-12 21:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98d586b2dc8897e787e430cccb4179231f302389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-12 22:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0f8cfbb3d2f46e0ae8bb2e0f5e7763e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c141bdcee8470caa8b8df8a9c56994 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a241623a6f64779aec99e2dd4b1785d : SUCCESS in 9m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/764712ea14034cea80b9eacddad2f9e9 : SUCCESS in 11m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59336b12798643baa89d55ace7a0a72d : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a303f29d03ea4e7f8888e8f4c67148c6 : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb2474bac0ce44f2aa879a668a8c1663 : SUCCESS in 18m 16s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7da71dbb84d64bfb98f5ad482183c058 : SUCCESS in 26m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/236a50984e27473d8b332d07dc1cec3f : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8980c8447a7b291eeb735069094f7108dd2cc582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-13 13:38:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"73e59aae6cefb07aaacaa2dfa8ba8a4d8fc5446d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 14:25:00.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c23d7b4eb8ce4a37a9eb3c00b2a7ab67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d066d2664474ee9adbb02f1e38563c1 : FAILURE in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00365cc33c944ead80bc2d7c2e119663 : FAILURE in 9m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e80bf79a2e1422ebba4ab1589879d2f : FAILURE in 11m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d44f827920ee4692987e247cce5bde74 : FAILURE in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d87db6e2ea1343218664f65f44b34d78 : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce21d7059c184b78b7f28ff681febd6d : FAILURE in 14m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/843a6d2121a747e8878668c846e5da76 : SUCCESS in 17m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20e783d350a94bb7b2222a08b9857a35 : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9ad1aa16b6a55606325a909bd5ca6521cded7695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-13 15:27:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5760fa71d640626ba66e2cc6a0b89cf2f86c9a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 16:11:21.000000000","message":"Patch Set 3: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff5f18f6cbe4457a169e405d57462c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56abcabda7d47d38aa0a3a9cca52c73 : FAILURE in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06dcc1fc79104e8c9cdca5b59a9bc5e1 : SUCCESS in 10m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c2dc2e3bb4a482abbdbae7096b604d6 : SUCCESS in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcd0287d003646078846200e432fadda : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4aed5bc07541ecaf6e4948860e61fa : SUCCESS in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2807a291c2e94b86a3720699369562b3 : SUCCESS in 16m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1d170b1625e440d387ecef043edcae76 : SUCCESS in 38m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc088352143b4900998aa459d272f86d : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6e1eefdc4846b61d593e4112aded4efdaa7b2ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-14 20:53:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"99515fab8564e952eafa581389df0463ca09b71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-14 20:54:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f57b5728ac968cb81928d7f3a808087c5a97dfc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-14 21:02:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"23cbbc9d18429a8ddc87d8c411b4c71913c6624b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 21:51:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b60fcee7d1f4922bf4ce9c5abd271bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7122890fd60496f8758d9516b29e49c : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d5074d3c3ff4897b5c9ad39315f8b3b : SUCCESS in 10m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c1817a22b284b4d8d83006dcdb5f396 : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65f77e9c2784de8b4c3d04330ccf7ac : SUCCESS in 8m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/533c76dce65e44ae8c19d96c8d7283e0 : SUCCESS in 14m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1c0db0ffd759478cb736a14c62f897a2 : SUCCESS in 37m 26s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/672980df62234539a7fb210c08f8cf67 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":6},{"id":"c4c79d388db71bc548aadc10d76e53ebb3618198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-15 10:41:31.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f075ad126b3c993b89c51311af94ee18177574b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 11:27:23.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa2e41075234fc68f11733bd5057bf1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062711f2eb47466c88d1aeca2c31a1c6 : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c2929f0dbe1445db24bb04e9362e3bd : FAILURE in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e66c010cb35941a6921643e430929bd1 : FAILURE in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eff3eaba4db40ada3c77ce1f1e902ab : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0a86db6808c4a5f989633793001de4f : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9d226843728412f9dc69fc21543aa1e : FAILURE in 15m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f5be720911ae4c459d54770f422ebec1 : SUCCESS in 39m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee740798ae84e5e8093540abdb0d362 : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":7},{"id":"d3e445fe5a3188256d1768195fe6ff09da9ce126","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-10-15 12:25:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b1e0cf83d3703f67790c25e91cbca80ad8ba8f02","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-15 12:26:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f5008d2463da27e9bca0edbbb7b306d3b0c735f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-15 12:52:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"39565e90b9c6903817531028e9f3872054701ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-15 13:29:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c63d1d7e1921e63e31b2e6c838b40a10c23f87bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-10-15 13:46:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c40579c5f1c4606f969a6312f5814cd89844b6a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 14:35:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac27d883adda4c06a240a03c58eab4e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e4928ab09b4da7a6c8698bf46e432b : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29181d0e19f94b3f8a5a10ba49cc13de : SUCCESS in 12m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea5a92dcb498457082f84da0d9546679 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375429a630a64e4a894a307d0473d2ea : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d3ef396156b4513adf3526a15841c3d : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd7f318ba3540c5a62c1f318ce58d2d : SUCCESS in 12m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0bd75006b84144d68af13305035d9a52 : SUCCESS in 44m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/006cb2f8d26647778dc3e87e68fa9d78 : SUCCESS in 24m 14s","accounts_in_message":[],"_revision_number":10},{"id":"3c7ad11efaa43d0a42eb8d1bcd53c456e6bf1e53","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-10-31 20:46:04.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"105201f29f2dc506a83ee4f77a9e087e638839b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 20:46:29.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"05e1f1f2421656aca6caa6c30dfad2d993e9b612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:33:22.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd404115af641aca06df80f1c6ead54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2257f8eea8341d9ae13dcbdbd47acd7 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e014336bcb5470d8ba6ec56ab8f99b7 : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de803381d29e40e49ebf5d7571c0cda1 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee60ca025f445948f10d623eb0f4bfc : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8371f6d86e0c4a338ea0f04d205ee513 : SUCCESS in 2m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/325e37611a9d4a5f80fdd51308e2cfe0 : SUCCESS in 29m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8465c2c0ccff47f5a764b0a170c4c655 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":10},{"id":"9091ab966b48b545328f1c122417e449d666a156","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:33:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"7a31065266d8adf6b4c31402e551ade5227ba90f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:34:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e25a391cc5e4940bde1c6adb7ebcc59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875a2ffaade847ec9e8a2629628be0d9 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e60073cbb974d6690cd4cf48c8d99c6 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"aedd2fef0adf8a50aac461f41923ce426378c533","revisions":{"4e52acbe1f9f09366dfc9cfee0bcffbd8dae759c":{"kind":"REWORK","_number":1,"created":"2025-10-12 21:35:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/1"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:35:03.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging. I\u0027d like to have the ssh\noutput instead, so I can tell what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility).\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e52acbe1f9f09366dfc9cfee0bcffbd8dae759c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e52acbe1f9f09366dfc9cfee0bcffbd8dae759c"}]},"branch":"refs/heads/master"},"4d79639df53abfec69919d6fadeccbb1502b3c1a":{"kind":"REWORK","_number":2,"created":"2025-10-13 13:38:43.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/2"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-13 13:36:15.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging. I\u0027d like to have the ssh\noutput instead, so I can tell what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility).\n\nThis patch also:\n- Add raise_when_error\u003dFalse parameter to prevent SSH utility\nfrom raising exceptions for command failures.\n- Add enhanced exception handling to catch SSH connection/transport\nfailures separately\n- Add proper result return even when return_result_on_error\u003dTrue for\nboth command failures and connection failures\n\nIn both case (with return_result_on_error\u003dTrue or False), the workflow\nexecusion still will be in error if the command fails, however, if\nreturn_result_on_error\u003dTrue the user will be able to read the command\nstdout, which wasn\u0027t the case previously.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d79639df53abfec69919d6fadeccbb1502b3c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d79639df53abfec69919d6fadeccbb1502b3c1a"}]},"branch":"refs/heads/master"},"e04667b31ab719e68df6c8b8faf4cb8e9a4e3339":{"kind":"REWORK","_number":3,"created":"2025-10-13 15:27:29.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/3"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-13 15:25:43.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging. I\u0027d like to have the ssh\noutput instead, so I can tell what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code. It\u0027s\ngoing to be up to the workflow to catch it.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e04667b31ab719e68df6c8b8faf4cb8e9a4e3339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e04667b31ab719e68df6c8b8faf4cb8e9a4e3339"}]},"branch":"refs/heads/master"},"3739bb36572df5df4f8d9702ceeec90d0ccf59f6":{"kind":"REWORK","_number":4,"created":"2025-10-14 20:53:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/4"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-14 20:46:29.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow, with something like this:\n\n  tasks:\n    execute_command:\n      action: std.ssh\n      input:\n        host: \"node1.example.com\"\n        username: \"root\"\n        private_key: |\n{{ ssh_private_key }}\n        cmd: \"/path/to/script\"\n        return_result_on_error: true\n      on-complete:\n        - collect_output\n\n    collect_output:\n      action: std.noop\n      publish:\n        ssh_stdout: \u003c% task(execute_command).result.stdout %\u003e\n        ssh_exit_code: \u003c% task(execute_command).result.exit_code %\u003e\n      on-complete:\n        - decision_task\n\n    decision_task:\n      action: std.echo\n      input:\n        output: \"Checking exit code...\"\n      on-success:\n        - log_completion: \u003c% $.ssh_exit_code \u003d 0 %\u003e\n        - fail_workflow: \u003c% $.ssh_exit_code !\u003d 0 %\u003e\n\n    fail_workflow:\n      action: std.fail\n\n    log_completion:\n      action: std.noop\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3739bb36572df5df4f8d9702ceeec90d0ccf59f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3739bb36572df5df4f8d9702ceeec90d0ccf59f6"}]},"branch":"refs/heads/master"},"dec577a1115ced4ebdbc8d398587c9e2f9856c60":{"kind":"REWORK","_number":5,"created":"2025-10-14 20:54:47.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/5"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-14 20:54:36.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow, with something like this:\n\n  tasks:\n    execute_command:\n      action: std.ssh\n      input:\n        host: \"node1.example.com\"\n        username: \"root\"\n        private_key: |\n{{ ssh_private_key }}\n        cmd: \"/path/to/script\"\n        return_result_on_error: true\n      on-complete:\n        - collect_output\n\n    collect_output:\n      action: std.noop\n      publish:\n        ssh_stdout: \u003c% task(execute_command).result.stdout %\u003e\n        ssh_exit_code: \u003c% task(execute_command).result.exit_code %\u003e\n      on-complete:\n        - decision_task\n\n    decision_task:\n      action: std.echo\n      input:\n        output: \"Checking exit code...\"\n      on-success:\n        - log_completion: \u003c% $.ssh_exit_code \u003d 0 %\u003e\n        - fail_workflow: \u003c% $.ssh_exit_code !\u003d 0 %\u003e\n\n    fail_workflow:\n      action: std.fail\n\n    log_completion:\n      action: std.noop\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dec577a1115ced4ebdbc8d398587c9e2f9856c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dec577a1115ced4ebdbc8d398587c9e2f9856c60"}]},"branch":"refs/heads/master"},"517146581f1585075462c75f3174bf85b372723e":{"kind":"REWORK","_number":6,"created":"2025-10-14 21:02:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/6"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-14 21:02:47.000000000","tz":120},"subject":"WIP: do not merge.","message":"WIP: do not merge.\n\nSSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow, with something like this:\n\n  tasks:\n    execute_command:\n      action: std.ssh\n      input:\n        host: \"node1.example.com\"\n        username: \"root\"\n        private_key: |\n{{ ssh_private_key }}\n        cmd: \"/path/to/script\"\n        return_result_on_error: true\n      on-complete:\n        - collect_output\n\n    collect_output:\n      action: std.noop\n      publish:\n        ssh_stdout: \u003c% task(execute_command).result.stdout %\u003e\n        ssh_exit_code: \u003c% task(execute_command).result.exit_code %\u003e\n      on-complete:\n        - decision_task\n\n    decision_task:\n      action: std.echo\n      input:\n        output: \"Checking exit code...\"\n      on-success:\n        - log_completion: \u003c% $.ssh_exit_code \u003d 0 %\u003e\n        - fail_workflow: \u003c% $.ssh_exit_code !\u003d 0 %\u003e\n\n    fail_workflow:\n      action: std.fail\n\n    log_completion:\n      action: std.noop\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/517146581f1585075462c75f3174bf85b372723e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/517146581f1585075462c75f3174bf85b372723e"}]},"branch":"refs/heads/master"},"c3e3679811f116621e3258cbae3e2f80719e03fa":{"kind":"REWORK","_number":7,"created":"2025-10-15 10:41:31.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/7"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-15 10:40:31.000000000","tz":120},"subject":"SSHAction: Add return_result_on_error\u003dTrue/False","message":"SSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow. Lookup in the doc change included in\nthis patch.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3e3679811f116621e3258cbae3e2f80719e03fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3e3679811f116621e3258cbae3e2f80719e03fa"}]},"branch":"refs/heads/master"},"8182b9b573cd1d95a37b3230cddbf6fd0ee73ad0":{"kind":"REWORK","_number":8,"created":"2025-10-15 12:52:01.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/8"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-15 12:51:49.000000000","tz":120},"subject":"SSHAction: Add return_result_on_error\u003dTrue/False","message":"SSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow. Lookup in the doc change included in\nthis patch.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8182b9b573cd1d95a37b3230cddbf6fd0ee73ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8182b9b573cd1d95a37b3230cddbf6fd0ee73ad0"}]},"branch":"refs/heads/master"},"12aa9a88f38e263de474210f8de83b931a9b3157":{"kind":"REWORK","_number":9,"created":"2025-10-15 13:29:23.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/9"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-15 13:29:12.000000000","tz":120},"subject":"SSHAction: Add return_result_on_error\u003dTrue/False","message":"SSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow. Lookup in the doc change included in\nthis patch.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/12aa9a88f38e263de474210f8de83b931a9b3157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/12aa9a88f38e263de474210f8de83b931a9b3157"}]},"branch":"refs/heads/master"},"aedd2fef0adf8a50aac461f41923ce426378c533":{"kind":"REWORK","_number":10,"created":"2025-10-15 13:46:06.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/14/963814/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/963814/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/963814/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/963814/10"}}},"commit":{"parents":[{"commit":"e2015bae38a8d1b0b3cf2453f51885cf4575db21","subject":"Merge \"Allow admin to retrieve everything\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e2015bae38a8d1b0b3cf2453f51885cf4575db21"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-12 21:30:47.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-10-15 13:45:54.000000000","tz":120},"subject":"SSHAction: Add return_result_on_error\u003dTrue/False","message":"SSHAction: Add return_result_on_error\u003dTrue/False\n\nCurrently, if an SSHAction is launched, and the target script returns\nwith non-zero, the execution will be set to ERROR, and we can see in\nthe \"execution output show\":\n\n\"The action raised an exception\n[action\u003d\u003cmistral.actions.std_actions.SSHAction object at 0x1234\u003e,\naction_ex_id\u003d\u003cUUID\u003e, msg\u003d\u0027Failed to execute ssh cmd\n\u0027/path/to/command param1 param2\u0027 on [\u0027hostname\u0027]\u0027]\"\n\nThat\u0027s not at all helpful for debugging, as the command stdout\nand stderr aren\u0027t accessible anymore. One may want to have the ssh\noutput instead, and see what went wrong.\n\nThis patch adds a return_result_on_error new param to have the\nfeature (and to keep backward compatibility). When this is set,\nwhen an SSH command is launched correctly, but returns non-zero,\nthe output will contain the stdout, stderr and return code.\nBecause of Mistral\u0027s design, the workflow execusion task cannot\nbe en ERROR, otherwise stdout/stderr/exit_code are overwritten\nby Mistral exception. Therefore, if one wants to still have the\nworkflow execution in error, the exit code must be looked-up\nlater in the workflow. Lookup in the doc change included in\nthis patch.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: If71f5663c26457a5cf6b9eb88f15355e664b520e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aedd2fef0adf8a50aac461f41923ce426378c533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/aedd2fef0adf8a50aac461f41923ce426378c533"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
