)]}'
{"id":"openstack%2Fmistral~981104","triplet_id":"openstack%2Fmistral~master~I3a3003831957c0b237dacecca49e29a15ba8e6dd","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-05-11 22:29:07.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-11 22:29:06.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-05-11 22:21:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I3a3003831957c0b237dacecca49e29a15ba8e6dd","subject":"Support OS_TEST_TIMEOUT","status":"MERGED","created":"2026-03-18 12:53:56.000000000","updated":"2026-05-11 23:07:01.000000000","submitted":"2026-05-11 23:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981104","meta_rev_id":"03bff9bb84a88e7b3f43e2d60d19dcd0e9794c82","_number":981104,"virtual_id_number":981104,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 23:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-11 22:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-05-11 22:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-05-11 22:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 13:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 19:17:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-30 19:17:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-05-11 22:29:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"15ac0b7d13b87edd81f90e31d48e5b51d0cd7854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-18 12:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff8aa67097377009da4bcbd4d5f7b4eb6834910f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:16:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59549fbeafa344e0a0d9347e04a147ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56c1e42b51d40ce9321bd107467a17f : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f49c559d9a124ff8b0f1edd20e27a9b2 : SUCCESS in 10m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce6ba1fa3aa54d31887b323673a824bb : SUCCESS in 10m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7108dda255764d509df20e156fd77a2e : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d7ab1dec9d48a19e494ada11dea122 : SUCCESS in 5m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3ee08d06a4c443ebb346a92f75d94e3 : SUCCESS in 15m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/56b067d41ad64baaacc0ef069f6ac447 : SUCCESS in 21m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a625dd9511534d4f859ca6ffa956eaeb : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"54c80ba99c26a5eff9b818fa64b00524fe8bd171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-30 16:43:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d40b7486bd8baa8140717075cf70c484b21d50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 17:10:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f474f34203f4cd895fe2e6f901d6329\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d8cd94905a444e8385dd7197de7d2e : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9dd70b7347c4847b419895fd0d23cf4 : SUCCESS in 10m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4af5f1f4a82245d8b5f981400a6df06a : SUCCESS in 11m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4223e999cffb4ea2ba0fe9cde22d6e90 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27772331c9d14a098de896097a38b78c : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e38dcae6777e47a7843bdd11de4fae86 : SUCCESS in 18m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3821b82786c74966a120ea24c4ef7b5c : SUCCESS in 18m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d29a49fab6494f2fadce2cd801e4e0d4 : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":2},{"id":"fef8c92416e6e0d05bafd1245715045d95cf08f9","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-11 22:21:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5bcca2e014f00c4cc161f27f6298e76487bd33d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 22:22:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"608e78c52355eb8aa798b8e1da660d7c386cdf24","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-11 22:29:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d2ba7b5bba30f2060b00da7607ffed33f43da85c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:05:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018b0514339949ab8c3d1a3240bff2bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611c76bb62f043ff9a31df2f93bde47c : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7da6af1d4ad149be97db55455241ae6b : SUCCESS in 9m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14cfba2f1f2248a298318e86c93f22bd : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6733d139af7944d3958b0014ba959440 : SUCCESS in 5m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2edc6008f5f04feaad492eabf06b8863 : SUCCESS in 17m 40s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e9cbcee2ce54de8a7fbda9d1b4e728e : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4885ab5cbf7d649648d8402d3853826a68fe24d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:05:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"03bff9bb84a88e7b3f43e2d60d19dcd0e9794c82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 23:07:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac13a1e8a914ce8a5598dce45a84002\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89621ac9badf43cdb6ac7b789a9108f1 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43254d9a7d6ee214a3d3d142864aab2fd998184b","revisions":{"8087bb164192c9e38e18589e03b6280525dc916e":{"kind":"REWORK","_number":1,"created":"2026-03-18 12:53:56.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/04/981104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/04/981104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/04/981104/1"}}},"commit":{"parents":[{"commit":"82d61ed56a1bae9b474093a73627e7fa40746bc3","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82d61ed56a1bae9b474093a73627e7fa40746bc3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-18 12:52:41.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-18 12:53:40.000000000","tz":60},"subject":"Support OS_TEST_TIMEOUT","message":"Support OS_TEST_TIMEOUT\n\nRunning parallel stestr, I get some unit tests that are stuck.\nTo fix this, I tried setting OS_TEST_TIMEOUT but it seems it\u0027s\nnot supported in Mistral. This patch adds support for it.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3a3003831957c0b237dacecca49e29a15ba8e6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8087bb164192c9e38e18589e03b6280525dc916e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8087bb164192c9e38e18589e03b6280525dc916e"}]},"branch":"refs/heads/master"},"43254d9a7d6ee214a3d3d142864aab2fd998184b":{"kind":"REWORK","_number":2,"created":"2026-03-30 16:43:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/981104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/04/981104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/04/981104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/04/981104/2"}}},"commit":{"parents":[{"commit":"82d61ed56a1bae9b474093a73627e7fa40746bc3","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/82d61ed56a1bae9b474093a73627e7fa40746bc3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-18 12:52:41.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-30 16:43:40.000000000","tz":0},"subject":"Support OS_TEST_TIMEOUT","message":"Support OS_TEST_TIMEOUT\n\nRunning parallel stestr, I get some unit tests that are stuck.\nTo fix this, I tried setting OS_TEST_TIMEOUT but it seems it\u0027s\nnot supported in Mistral. This patch adds support for it.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I3a3003831957c0b237dacecca49e29a15ba8e6dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43254d9a7d6ee214a3d3d142864aab2fd998184b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43254d9a7d6ee214a3d3d142864aab2fd998184b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
