)]}'
{"id":"openstack%2Fmistral~983430","triplet_id":"openstack%2Fmistral~master~I96877baad09446f15959a82deaac19a743bb376a","project":"openstack/mistral","branch":"master","topic":"assertFail","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-12 09:33:32.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-05-12 08:42:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I96877baad09446f15959a82deaac19a743bb376a","subject":"Replace wrong assertion method","status":"MERGED","created":"2026-04-06 13:45:07.000000000","updated":"2026-05-12 09:34:27.000000000","submitted":"2026-05-12 09:33:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"983430-assertFail","meta_rev_id":"3bea3abb41acfdb07fb35234d101ba2bdc9df559","_number":983430,"virtual_id_number":983430,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 09:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 08:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-12 08:42:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-06 14:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-06 18:17:20.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2026-05-12 08:42:21.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"4db0925682e0e1134bbcd45cd83f23a6f5a3b113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-06 13:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4166884936e966ad2465515c3ab3292fd7bd2a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 14:40:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f969223b51294c2194d1ffd1392f152b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eadccdc2dc4480f900922da9d035514 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf56748d8b4a4a44b4cad09ae5999495 : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7b1373e34394d9693f5719bcc8f9a18 : SUCCESS in 9m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6376eccc8a154b4498775fde78de9a17 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e283b6db9a0c424bb84cd32bbe0af86b : SUCCESS in 7m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2e13a7873e406a9d188164e713f035 : SUCCESS in 11m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/958a957286814e6fbc864bc80f15eaa2 : SUCCESS in 41m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b53d65ad55249e488dd59059e6063c1 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e23bd5f36e38e6c0b22b57937fd43c46f5510632","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-04-06 18:17:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f37a53ce1a32518ddbc2e0956f115ae1337f592f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-27 10:57:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c21bf05e59cecdd75e3117a9402d45b25bacdae","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-11 22:15:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9f1efd3c9f24c8db7f6c528da03d89606232899","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-12 02:34:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14df8e49f9c1680405c25f5910eeda33c32f019d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-12 08:42:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a90a58e4054e2bb8e741bde90d8faddb5eb46cda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 08:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c559db2e8392376c21d1ac765ada50463d5a92bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:33:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b398e0f86a9e4a6d9871a2618821b9d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c93dc78b5e44018b0f81e1117e0a1fb : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ade839a5d91447f3aa8b0d26e539073f : SUCCESS in 10m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70f52fa1cd3b431f9ecc2023b0628ae4 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c35b295271f43fdb223245f0e3999fc : SUCCESS in 9m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/528e6175c0e04ebf893f406144eeea21 : SUCCESS in 45m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9124ff5f460340c69a262cb727bd2688 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ce466cb2ada6848d689861f6b69e61156d804332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:33:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3bea3abb41acfdb07fb35234d101ba2bdc9df559","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 09:34:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492b79d93b4c472ab2c0e4fd47741633\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c918b36a0eeb4bd988bdd63b1bef3f27 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0abe4995a5a49c527d033d4f86c69a24b86dec6","revisions":{"d0abe4995a5a49c527d033d4f86c69a24b86dec6":{"kind":"REWORK","_number":1,"created":"2026-04-06 13:45:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/983430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/30/983430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/983430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/983430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/30/983430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/30/983430/1"}}},"commit":{"parents":[{"commit":"40f6a9eaab94b4245aa4b40caa980aae22f9d4be","subject":"Merge \"Drop redundant overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40f6a9eaab94b4245aa4b40caa980aae22f9d4be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-06 13:44:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-06 13:44:57.000000000","tz":540},"subject":"Replace wrong assertion method","message":"Replace wrong assertion method\n\nThe assertFail method does not exist.\n\nChange-Id: I96877baad09446f15959a82deaac19a743bb376a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0abe4995a5a49c527d033d4f86c69a24b86dec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d0abe4995a5a49c527d033d4f86c69a24b86dec6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
