)]}'
{"id":"openstack%2Fmistral~989875","triplet_id":"openstack%2Fmistral~master~I1cda758f0a73bb668fee82065c08cfc5705b5a8b","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-24 19:02:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1cda758f0a73bb668fee82065c08cfc5705b5a8b","subject":"Mark checksum calculation for no security","status":"MERGED","created":"2026-05-24 04:12:42.000000000","updated":"2026-05-24 19:03:39.000000000","submitted":"2026-05-24 19:02:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989875","meta_rev_id":"fc5ecb6cda174e161b333a56340b2ecd9e0ec6b4","_number":989875,"virtual_id_number":989875,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-24 19:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-24 18:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-24 18:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-24 04:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-24 18:20:08.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"bbf261816e728c9c44bede87321bef02af3c13e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-24 04:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8452ca037a735ee9c1300199da9f47608d1347e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 04:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23c5596a4449461bbc45a6a940ae884b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf03c55a6c840ca9a416cd8cbb5121b : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddf6ad747d124aab85815f6dcdb9eaa4 : SUCCESS in 8m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9412d523a2034c3ea518aaf5841c02ab : SUCCESS in 9m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8588552e6ebf4e6c814430d61e4d9df7 : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498938a750a34b9a93fad02d4a714d82 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb088ba60253424e92efde736af8564d : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d96a3a35532484d9891d1634021b1d6 : SUCCESS in 14m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3328dc7992c44df48f2d0584c6c0b42b : SUCCESS in 41m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/495b63caa76b4ac880b14f8bca8c2626 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":1},{"id":"de0cd04326509011822a616ff56ac9fd72c793c7","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-24 18:20:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce78379ce83a3b0dd1c0756e6bbedc427a8f5bdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 18:20:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bf9a82681053e2e1a24760482132ec5d9896dd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 19:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17033400649a4494b0bc79631bf8cde8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85033dd7444e496bbe15537550151d6d : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/126438c11941466d99431479ca44313b : SUCCESS in 8m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a0396fa5eb848f3b90236fc2071254c : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cbfd17604842a9be3d5c03eaceeaa4 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4c58ea5b1d0496abd949d45e379a834 : SUCCESS in 4m 05s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38da914e00d74d6fa27697dbd65a3f81 : SUCCESS in 37m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b580cb15ba4f2c8f22887cec5fe4e6 : SUCCESS in 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cca609cfde7501aba232f581d4360f07272446b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 19:02:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fc5ecb6cda174e161b333a56340b2ecd9e0ec6b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 19:03:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778c22f80d454791b7d35b1330c6455f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0794282c7e4161a8213240b6b22ca8 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c029bb08640179682f657ac5dc910f6034fa3d5e","revisions":{"c029bb08640179682f657ac5dc910f6034fa3d5e":{"kind":"REWORK","_number":1,"created":"2026-05-24 04:12:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/989875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/75/989875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/989875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/989875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/75/989875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/75/989875/1"}}},"commit":{"parents":[{"commit":"9e0877a9a6dd15544490877d4b85a2d5dae3f302","subject":"Merge \"Replace own exception message assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9e0877a9a6dd15544490877d4b85a2d5dae3f302"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 04:11:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-24 04:11:44.000000000","tz":540},"subject":"Mark checksum calculation for no security","message":"Mark checksum calculation for no security\n\nhashlib.md5 is not allowed in FIPs mode unless it\u0027s explicitly marked\nwith usedforsecurity\u003dTrue .\n\nChange-Id: I1cda758f0a73bb668fee82065c08cfc5705b5a8b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c029bb08640179682f657ac5dc910f6034fa3d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c029bb08640179682f657ac5dc910f6034fa3d5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
