)]}'
{"id":"openstack%2Fmistral~989960","triplet_id":"openstack%2Fmistral~master~Ie8d30b5fe11edf5483250b20aa7a8960c1337ed4","project":"openstack/mistral","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-05-25 17:38:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ie8d30b5fe11edf5483250b20aa7a8960c1337ed4","subject":"Remove management_url dead code","status":"MERGED","created":"2026-05-25 13:11:30.000000000","updated":"2026-06-06 09:53:31.000000000","submitted":"2026-06-06 09:52:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989960","meta_rev_id":"967c4ca516feb5ab1634840954bc6194ae13abfb","_number":989960,"virtual_id_number":989960,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-06 09:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-25 17:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-06 08:57:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 13:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d628825435bb4b84af710b0fb27b78f2b3dcc39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-25 13:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63b4955554936bc6c68ee9204bd00d5abb54e9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 13:58:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6028058c697445baa23b98267325a797\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0430359fba5d48369d7b3bc32142860b : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0644bf98e66d40cf8e2f5949426df2c4 : SUCCESS in 10m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/124e964cad594a2680d207134f8fdf2e : SUCCESS in 9m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d530ce2cd81498a91456357ce60e74f : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63689f663cc41d8999b325650d93894 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75926858961d4e838fa94f8688508642 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5051d7a4d474973bcbc3d7abcebd150 : SUCCESS in 18m 21s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/36cbda2755214d69937598514369e2fe : FAILURE in 40m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39c77ffa3d0e4aa8859f3387f76fd4b5 : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"71af803c6d45c08cb1edfa1ef4bdb364901329af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-25 15:36:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e6535069e454586fde697e50983eba249e62d3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 16:23:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9e6636d38d47b2a96b4ee3183f9ca6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0978b5da3f854317a7ec9d0124019e12 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ba21d4d179a49e085d8915d48be5590 : SUCCESS in 8m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/153ce38dd1cc4419a29b76afc55bb211 : SUCCESS in 9m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6dcf7866e9074967a4817862207e2f0a : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a184744434ce4dfc96baa2be180864da : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5323e85b152048fcb00a075b53414504 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819c39a0320746b4ad49db77000c0cbc : SUCCESS in 12m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8581e5d9d22e4b2b89b1a7058f314ea0 : SUCCESS in 40m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e25ef983e374afc8a420f070b796a46 : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4309161340c4f12fb222a30fb0b80ff51ad61510","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-25 17:38:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f35d0d1ff05ce86753844f4e7b082fc607c32d0d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-06-06 08:57:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d870564c89e22a640814e76ed8e035208f360271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 08:58:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fae8354fbae83c8e9e91cfe97bf9e1d3a364a81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 09:52:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f245cb9193e34171b19c5618cc03d8e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f075551b18d468ba2de9e0ff9b0e2c9 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a795a708f2104779a2e79bca4642c93e : SUCCESS in 9m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6e4b4d3daf1462cb285b94100165642 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bc322bb31d41eaa0c38a416fbac2ea : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87828807049c468ea2cf903c4801fe05 : SUCCESS in 3m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b78e92fd01e04644bff4b20d10ce8ec9 : SUCCESS in 38m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333851b7d12f49cfbcdd8480ac7647fd : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6a9df832d882051725cbd1794c6018eeefe7522b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 09:52:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"967c4ca516feb5ab1634840954bc6194ae13abfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 09:53:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb020c65b2ac4e82bbcfb7bd9ad931c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa0bc22d3a54467a177d1802128b188 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4aaecb58c165dc6a81ec971e5ac182028d2ec1d","revisions":{"4f67ca1bd4e5c9fcb662a744781162a0a8ebe460":{"kind":"REWORK","_number":1,"created":"2026-05-25 13:11:30.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/60/989960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/989960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/989960/1"}}},"commit":{"parents":[{"commit":"ecdccbd55181d963ce6ff97281722de2cac851c9","subject":"Fix trust-scoped token auth failure in keystone client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecdccbd55181d963ce6ff97281722de2cac851c9"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-05-25 13:11:08.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-05-25 13:11:08.000000000","tz":120},"subject":"Remove management_url dead code","message":"Remove management_url dead code\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nChange-Id: Ie8d30b5fe11edf5483250b20aa7a8960c1337ed4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f67ca1bd4e5c9fcb662a744781162a0a8ebe460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4f67ca1bd4e5c9fcb662a744781162a0a8ebe460"}]},"branch":"refs/heads/master"},"c4aaecb58c165dc6a81ec971e5ac182028d2ec1d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-25 15:36:44.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/60/989960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/989960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/989960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/989960/2"}}},"commit":{"parents":[{"commit":"b9e5636707c97e6474decae67571e9f6f32514a9","subject":"Fix trust-scoped token auth failure in keystone client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9e5636707c97e6474decae67571e9f6f32514a9"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-05-25 13:11:08.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-05-25 15:34:27.000000000","tz":120},"subject":"Remove management_url dead code","message":"Remove management_url dead code\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nChange-Id: Ie8d30b5fe11edf5483250b20aa7a8960c1337ed4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4aaecb58c165dc6a81ec971e5ac182028d2ec1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c4aaecb58c165dc6a81ec971e5ac182028d2ec1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
