)]}'
{"id":"openstack%2Fmistral~991544","triplet_id":"openstack%2Fmistral~stable%2F2025.1~Ide923b606df884cb7174e2c6c0d094b76c94e893","project":"openstack/mistral","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-06-04 12:04:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide923b606df884cb7174e2c6c0d094b76c94e893","subject":"Disable failing jobs","status":"ABANDONED","created":"2026-06-04 07:52:41.000000000","updated":"2026-06-04 12:04:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ddd3756e87c1427c71d6183982a1f3b3322b4d6d","_number":991544,"virtual_id_number":991544,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-04 09:32:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 09:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2ecd65f643522ce1cfdabdeca143e5c544a60b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-06-04 07:52:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1c86955e252071e8c5723805d6a1b2dd3cc697f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-06-04 08:38:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d23b8961bc2ff5ca559c9f99377015ee468a0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:32:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73d72f7b554d431186d5c9483c5ee0b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d2a0833433480f9d24d477c8e98c2c : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ff1069b7cd4b20b0385c811946830e : FAILURE in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4444d16993d14dd1b11643a23a027e5a : FAILURE in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48607c7f287487f92a679c3a2e6c82a : FAILURE in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e95c35dac970450482444f73b3fde1de : SUCCESS in 4m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/af62f55a13484138bc37a8fac518410b : SUCCESS in 18m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39ba3d5f3268445eb9a390bf529387e2 : SUCCESS in 41m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/890547260dc8498b952e6b2c87b0ec88 : SUCCESS in 40m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/863e17fe169e4da188dc988c7b0be125 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ddd3756e87c1427c71d6183982a1f3b3322b4d6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-06-04 12:04:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b65dd5f31b4b2ac461b268cb487a8dfba15a6d13","revisions":{"340b393dcb10098c894d70879bb54408cf1362c5":{"kind":"REWORK","_number":1,"created":"2026-06-04 07:52:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/44/991544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/991544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/991544/1"}}},"commit":{"parents":[{"commit":"c79e903f495e7b85f9713befd5a945fb71efde2b","subject":"Do not call setup_db from engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c79e903f495e7b85f9713befd5a945fb71efde2b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-06-04 07:50:51.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-06-04 07:52:07.000000000","tz":120},"subject":"Disable failing jobs","message":"Disable failing jobs\n\nAs of 2026, we cannot test anymore on mistral-devstack-jammy and\nmistral-devstack-non-apache-tempest-ipv6-only because of setuptools\nupgrades.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nChange-Id: Ide923b606df884cb7174e2c6c0d094b76c94e893\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/340b393dcb10098c894d70879bb54408cf1362c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/340b393dcb10098c894d70879bb54408cf1362c5"}]},"branch":"refs/heads/stable/2025.1"},"b65dd5f31b4b2ac461b268cb487a8dfba15a6d13":{"kind":"REWORK","_number":2,"created":"2026-06-04 08:38:06.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/44/991544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/991544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/991544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/991544/2"}}},"commit":{"parents":[{"commit":"c79e903f495e7b85f9713befd5a945fb71efde2b","subject":"Do not call setup_db from engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c79e903f495e7b85f9713befd5a945fb71efde2b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-06-04 07:50:51.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","date":"2026-06-04 08:24:20.000000000","tz":120},"subject":"Disable failing jobs","message":"Disable failing jobs\n\nAs of 2026, we cannot test anymore on mistral-devstack-jammy and\nmistral-devstack-non-apache-tempest-ipv6-only because of setuptools\nupgrades.\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nChange-Id: Ide923b606df884cb7174e2c6c0d094b76c94e893\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b65dd5f31b4b2ac461b268cb487a8dfba15a6d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b65dd5f31b4b2ac461b268cb487a8dfba15a6d13"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
