)]}'
{"id":"openstack%2Fmonasca-agent~554304","triplet_id":"openstack%2Fmonasca-agent~master~I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe","project":"openstack/monasca-agent","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe","subject":"Removed dependency on supervisor","status":"MERGED","created":"2018-03-19 17:40:43.000000000","updated":"2018-06-29 14:20:42.000000000","submitted":"2018-06-29 14:20:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"554304-1530282042217-4363ebbb","meta_rev_id":"30bfdcdf76ffe90ff1816e7df4cfb6392ceff8d6","_number":554304,"virtual_id_number":554304,"owner":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":2,"date":"2018-06-29 14:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-29 12:50:53.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2018-06-28 09:43:23.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2018-06-27 11:08:33.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":1,"date":"2018-06-28 09:43:23.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-29 15:09:30.000000000","updated_by":{"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"},"reviewer":{"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"},"state":"REVIEWER"},{"updated":"2018-04-10 22:01:59.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2018-05-07 12:06:32.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2018-06-27 08:05:40.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2018-06-27 11:08:33.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2018-06-28 09:43:23.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2018-06-29 12:50:53.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2018-06-29 14:20:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f8dec493c06c1fab537812d4aaf13b814151c7c","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-19 17:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d570cb9cc2506ec20ca55b2efdbbacd617329ecb","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-19 17:42:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d64ef4ba98753b11f0e3f0232f103fb5dd044189","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-19 18:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/1/check/openstack-tox-pep8/e72c2e2/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/1/check/openstack-tox-py27/ce25050/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/1/check/openstack-tox-py35/24c6ff2/ : SUCCESS in 4m 09s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/1/check/monasca-tempest-python-influxdb/88c34c7/ : FAILURE in 34m 14s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/1/check/monasca-tempest-java-influxdb/c604e4c/ : FAILURE in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7b4b8e8cab5afce183d3fbada597923d1d9e84e","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-19 21:32:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e18975019a3aff567c19a34430f50c21df5b9953","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-19 22:03:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/2/check/openstack-tox-pep8/3aeee40/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/2/check/openstack-tox-py27/3046793/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/2/check/openstack-tox-py35/6c8430c/ : SUCCESS in 4m 11s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/2/check/monasca-tempest-python-influxdb/c130caa/ : FAILURE in 30m 02s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/2/check/monasca-tempest-java-influxdb/2d3a6cc/ : FAILURE in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c21cb9ead99b2dc9160f2135aada2f102bfc7120","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-22 17:22:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"307f081f1ce09c32de9a942458520bc00d044737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 17:58:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/3/check/openstack-tox-pep8/b544590/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/3/check/openstack-tox-py27/11f2b42/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/3/check/openstack-tox-py35/60dcd56/ : SUCCESS in 3m 57s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/3/check/monasca-tempest-python-influxdb/8f496a7/ : FAILURE in 34m 05s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/3/check/monasca-tempest-java-influxdb/2d88235/ : FAILURE in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ae19ebd381e99c206628d9e002d13ba1d04db658","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-03-23 15:14:17.000000000","message":"Patch Set 3: Code-Review-1\n\nYou also have to update [files] section in setup.cfg [1].\n\n[1] https://git.openstack.org/cgit/openstack/monasca-agent/tree/setup.cfg#n19","accounts_in_message":[],"_revision_number":3},{"id":"985ba622ce7a4da8bbab40e0a557f02439678c62","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-03-23 15:15:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efd9f4cad39b6d2966e6a93679f5124b59d79088","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-23 22:14:39.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f0d3359f9b5cf7a668d57e2f0ee5d0ef64303251","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-23 22:17:09.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"faefda2513b3eccb11cdff48796487f313229f82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 22:51:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/5/check/openstack-tox-pep8/7c2aa0a/ : SUCCESS in 5m 17s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/5/check/openstack-tox-py27/dc5930f/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/5/check/openstack-tox-py35/259ccff/ : SUCCESS in 4m 47s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/5/check/monasca-tempest-python-influxdb/5d03af0/ : FAILURE in 31m 37s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/5/check/monasca-tempest-java-influxdb/5c0959c/ : FAILURE in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c0c0cd57ebe85ee376879c3c24e5231befc6f61c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-03-28 11:52:05.000000000","message":"Patch Set 5: Code-Review-1\n\nPlease see my comment to PS3.","accounts_in_message":[],"_revision_number":5},{"id":"ac3d04d2b575a1a61467e77ca2670355b86cfaa3","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-03-28 20:23:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e60587d6de6416d53895920d8a6e8510893c007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 21:04:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/6/check/openstack-tox-pep8/d79dbde/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/6/check/openstack-tox-py27/bd34fe0/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/6/check/openstack-tox-py35/a31e738/ : SUCCESS in 4m 13s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/6/check/monasca-tempest-python-influxdb/53bc687/ : FAILURE in 35m 25s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/6/check/monasca-tempest-java-influxdb/e5dce62/ : FAILURE in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0a8a63f9f1434b9b23a55c50a852d10e066057dd","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-03-28 21:15:55.000000000","message":"Patch Set 6:\n\nLooks good.  Should each of the .service.template files have the same Description?  Seems like you could add the service name in parenthesis, but that is just cosmetic.","accounts_in_message":[],"_revision_number":6},{"id":"bc5f80830d6b8116e269c850fb5f8b644830d2d8","author":{"_account_id":26639,"name":"Tobias Johansson","email":"tjohansson@op5.com","username":"tobiajo"},"date":"2018-03-29 15:09:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"3557ad86faf0831cea17f2f0ecaa5bce56bb8ac7","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-04-10 20:56:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98167cd5172ca68abb740db3c521e8f2f188ef05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 21:37:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/7/check/openstack-tox-pep8/eb80b65/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/7/check/openstack-tox-py27/ce84cd5/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/7/check/openstack-tox-py35/cc19dce/ : SUCCESS in 4m 45s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/7/check/monasca-tempest-python-influxdb/90a5966/ : FAILURE in 34m 13s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/7/check/monasca-tempest-java-influxdb/101989a/ : FAILURE in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5ef1eb515047a106c5433cfa0d9d9ebaef317e84","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-04-10 22:01:59.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5376734e432efb1ca0287b712aaa7afebfdda17c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-04-11 11:48:49.000000000","message":"Topic changed from 2to3 to goal-python35","accounts_in_message":[],"_revision_number":7},{"id":"da01e22055df5f3664374d32c6b6ad7926305073","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-04-11 11:56:55.000000000","message":"Patch Set 7:\n\nIt breaks devstack deployment [1].\n\n[1] http://logs.openstack.org/04/554304/7/check/monasca-tempest-python-influxdb/90a5966/logs/devstacklog.txt.gz#_2018-04-10_21_36_18_611","accounts_in_message":[],"_revision_number":7},{"id":"8a30b0dd670bfab53a4c0321d330b5cdca563483","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-04-11 15:26:58.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nNote to self","accounts_in_message":[],"_revision_number":7},{"id":"6b793b2d0c831fb664865dd3c2c7f63ddfd083c1","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-04-11 20:34:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a970365c89a65cbdba39a224c22da3635944f983","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-04-11 21:24:41.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"43669fe90668a37d7e9efa63c4afe159b9668919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 01:02:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/9/check/openstack-tox-pep8/304ba66/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/9/check/openstack-tox-py27/104f90c/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/9/check/openstack-tox-py35/15a5c82/ : SUCCESS in 4m 15s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/9/check/monasca-tempest-python-influxdb/76294c2/ : FAILURE in 33m 32s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/9/check/monasca-tempest-java-influxdb/f3a6fa2/ : FAILURE in 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5edf454c969547cdb1a14e24a8bf59884c6b57f8","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-02 15:10:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c2fa30f7abc03c6abd0c8d7f1e14bcb4489e38ba","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-02 15:11:27.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"af7400d09b5f5bcebc6486a4b9143ad9773db72a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 15:56:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/10/check/openstack-tox-pep8/81b6bb0/ : FAILURE in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/10/check/openstack-tox-py27/2fdae71/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/10/check/openstack-tox-py35/6f4494a/ : SUCCESS in 4m 06s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/10/check/monasca-tempest-python-influxdb/253e5da/ : FAILURE in 32m 07s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/10/check/monasca-tempest-java-influxdb/761eb6b/ : FAILURE in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"96bb92ec61a5a0bb9506a04d970a0735ffa8139e","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-03 01:06:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3a07a0f45a5277083ef0978bebfc1556db92199b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 01:41:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/11/check/openstack-tox-pep8/1304559/ : FAILURE in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/11/check/openstack-tox-py27/3107a32/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/11/check/openstack-tox-py35/9233517/ : SUCCESS in 4m 34s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/11/check/monasca-tempest-python-influxdb/ac543fa/ : FAILURE in 34m 27s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/11/check/monasca-tempest-java-influxdb/ed404da/ : FAILURE in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aab1862fc1ff6682575c418c5bf583d8170ade41","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-04 22:36:47.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3111c7eced4447eaf88bc36c4840593b0594607c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-04 23:25:49.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/12/check/openstack-tox-pep8/6bb4bd6/ : FAILURE in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/12/check/openstack-tox-py27/9b6346e/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/12/check/openstack-tox-py35/a9b0004/ : SUCCESS in 3m 17s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/12/check/monasca-tempest-python-influxdb/99d82b2/ : FAILURE in 35m 07s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/12/check/monasca-tempest-java-influxdb/8deac3b/ : FAILURE in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"41e61ef6dbf1bbd912936565d8d09965070e6541","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-06 20:46:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0c46921d168fbc89972d33f17338b15bd3f13cfd","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-06 21:06:05.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"aa80fcb80c241c7109121650fa8160dfc87ac64b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-06 21:47:34.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":14},{"id":"9311d75c8bfc937c8a4887ef91d5fa87c5b4b20c","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2018-05-07 12:06:32.000000000","message":"Patch Set 14: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":14},{"id":"7b57866be60db9a6d4db5c75b3eec321413b474b","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-07 14:43:27.000000000","message":"Patch Set 14:\n\n(10 comments)\n\nThanks very much, Dobroslaw.","accounts_in_message":[],"_revision_number":14},{"id":"45a6e3eec43b3885288840f642889a3b0b7c743d","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-07 14:43:41.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"799bf16a9bf3b89533360aaebccd4b4467aef3c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 14:58:28.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"0edc7b6721eea0dff53cca9ba9c2fd1ec600f48a","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-07 15:28:25.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"fd95520d0890e13377a1066f2f6886ce642200a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 16:37:42.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/16/check/openstack-tox-pep8/f361c04/ : FAILURE in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/16/check/openstack-tox-py27/c8a9dc7/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/16/check/openstack-tox-py35/371ac20/ : SUCCESS in 4m 25s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/16/check/monasca-tempest-python-influxdb/f7a8332/ : SUCCESS in 1h 01m 06s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/16/check/monasca-tempest-java-influxdb/3901257/ : SUCCESS in 1h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4c05fc6cc1d86758268c5cf17b9bcde5c50ae14c","author":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"date":"2018-05-07 19:35:28.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5724f437a193c4ee60c5d8d549d0ce10554f1828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-07 20:27:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/17/check/openstack-tox-pep8/51574dd/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/17/check/openstack-tox-py27/1a5cf1c/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/17/check/openstack-tox-py35/72467e3/ : SUCCESS in 5m 39s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/17/check/monasca-tempest-python-influxdb/8936339/ : FAILURE in 10m 50s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/17/check/monasca-tempest-java-influxdb/24f54af/ : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1dd848797b23f2f5b18d20c35446bfc420625119","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-05-08 14:38:08.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"c7d1c66c719b409263af1d57c52f8123340c5d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 15:38:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/17/check/openstack-tox-pep8/d720542/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/17/check/openstack-tox-py27/1e530c4/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/17/check/openstack-tox-py35/99d2df8/ : SUCCESS in 4m 05s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/17/check/monasca-tempest-python-influxdb/e536fd7/ : SUCCESS in 42m 40s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/17/check/monasca-tempest-java-influxdb/2681725/ : SUCCESS in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7d867eb28c3eb49af180b82dcd5d5d523afe8a96","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-05-11 13:54:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3f2bc64634ef3eccbf5ede89b3db7353952ac894","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-05-17 11:18:45.000000000","message":"Patch Set 17: Code-Review-1\n\n(10 comments)\n\nThanks for this. Looks like a good start. A grep on the code base shows a fair bit of cruft floating around for supervisord. For example in the documentation there are still references to it, and in monasca-agent/packaging. Some comments inline.","accounts_in_message":[],"_revision_number":17},{"id":"c33f071c542594ed585392c1af25b34dee1cd12a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-05-17 11:26:15.000000000","message":"Patch Set 17:\n\nAre we able to add some units tests for the new Python code?","accounts_in_message":[],"_revision_number":17},{"id":"f9a1c28d38a933b2fa8dc53f9b2740190b1faf92","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-12 13:41:03.000000000","message":"Patch Set 17:\n\nAny progress on this?","accounts_in_message":[],"_revision_number":17},{"id":"13562f41cac6622295f11f0dbc7b723e17e0c0dc","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-13 09:04:52.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"c61d54a10007c07931b9742b9252c19f84139051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 10:22:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/18/check/openstack-tox-pep8/7fdf016/ : SUCCESS in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/18/check/openstack-tox-py27/92739ef/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/18/check/openstack-tox-py35/9e95d74/ : SUCCESS in 5m 50s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/18/check/monasca-tempest-python-influxdb/598576c/ : SUCCESS in 46m 32s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/18/check/monasca-tempest-java-influxdb/9f0fdcc/ : SUCCESS in 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6087b0a617ba04de7166f73b0a2e946283f210e6","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 08:33:50.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"ba98521134bd4be7c93c10c8555884f0c22a4cc8","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 08:35:52.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6787b8f651298c7112b89fdab2e780b8ce04d45a","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 08:43:51.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"8233dd52c55140bef9263a1e45f1a1be0fc3cd03","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 08:47:32.000000000","message":"Patch Set 21:\n\n(8 comments)","accounts_in_message":[],"_revision_number":21},{"id":"12f95ffbe5b43f07abaa457c403ffd748c27b847","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 09:11:43.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"a5f3f31dc627c1f87a7a1e052a84a6c009441b13","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-06-25 09:19:26.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)\n\nThanks Adrian.\n\nI still see some rogue references to supervisord, for example in docs/Agent.md.","accounts_in_message":[],"_revision_number":22},{"id":"4992a7590d6964c727ce4eac91c1471311f5f5a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 10:14:48.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/22/check/openstack-tox-pep8/2baa552/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/22/check/openstack-tox-py27/0715a43/ : SUCCESS in 6m 22s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/22/check/openstack-tox-py35/24c3dff/ : SUCCESS in 5m 17s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/22/check/monasca-tempest-python-influxdb/01d7862/ : SUCCESS in 53m 07s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/22/check/monasca-tempest-java-influxdb/68821cc/ : SUCCESS in 49m 26s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"edcefb8bbf64a7a2d4b652faeb53e845cbd871de","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 13:31:16.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8e08c53a7ebe742fb331f811f390595bf0f30e6f","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 13:32:55.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"16f9c9e3cf854a11e3f9da3ebf3e494ab985669b","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-25 13:33:38.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"dfa1b3eb4ef0198ea20fea61a58039cad9e71843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 14:14:25.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/23/check/openstack-tox-pep8/d188a75/ : FAILURE in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/23/check/openstack-tox-py27/c710b0e/ : FAILURE in 5m 11s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/23/check/openstack-tox-py35/f256e5c/ : SUCCESS in 4m 43s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/23/check/monasca-tempest-python-influxdb/c1e9060/ : FAILURE in 32m 34s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/23/check/monasca-tempest-java-influxdb/0070e71/ : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f78c514ec2e48359b9435f53490037b0080fb365","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-26 06:42:45.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"3910e57eeeac40a7cb55cb1fffe8c4274091026d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-26 07:58:14.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/24/check/openstack-tox-pep8/b22cce6/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/24/check/openstack-tox-py27/cdf90e3/ : SUCCESS in 6m 44s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/24/check/openstack-tox-py35/cad046e/ : SUCCESS in 6m 30s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/24/check/monasca-tempest-python-influxdb/12c4100/ : SUCCESS in 49m 40s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/24/check/monasca-tempest-java-influxdb/e719395/ : SUCCESS in 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8357ddb24fb6c95065795f13c9fe75dc76b9b360","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-26 12:56:28.000000000","message":"Patch Set 24:\n\n(6 comments)\n\nSmall comments, other than that looks good.","accounts_in_message":[],"_revision_number":24},{"id":"e199ee5c0e158cfd23ae331e7c4ea911b9271c8d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-26 14:05:40.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"07e4ab1c9cd8358b333ef14f21d05ff9e3f76027","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-27 08:47:44.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"58463dccedbf34eb4252ea8ac9523f8dc0b2dac5","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-27 08:48:59.000000000","message":"Patch Set 24:\n\n(7 comments)","accounts_in_message":[],"_revision_number":24},{"id":"76ace2b233e99e5c8c6e5915bc2251b054a832ab","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-27 09:23:50.000000000","message":"Patch Set 25: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"f72f1f18578398752f03e47adea28bbaeb365d20","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-27 09:28:53.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"1f4b3c986ce40d68e1364ba387f6d01e30a28b93","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-27 09:29:57.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"5acdb16574d0ee2040e466926b619033b3e9909a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-27 11:02:30.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/26/check/openstack-tox-pep8/4253071/ : SUCCESS in 6m 33s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/26/check/openstack-tox-py27/3663a6c/ : SUCCESS in 7m 09s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/26/check/openstack-tox-py35/21fe072/ : SUCCESS in 4m 39s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-python-influxdb/9bd397b/ : SUCCESS in 41m 50s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-java-influxdb/a261c57/ : SUCCESS in 49m 07s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"db2afb8f38d878c0fa4634722002a811a6e20d0d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-27 11:08:33.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"a8191d305f71f1f1754015f7469da2e108687655","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-06-27 16:35:11.000000000","message":"Patch Set 26: Code-Review-1\n\nThere is still a lot of supervisord related stuff in packaging which would be nice to remove to reduce confusion:\n\n```\ndoug@xps15:~/git/monasca-agent$ grep -ri supervisor\npackaging/monasca-agent.init.template:PATH\u003d$PATH:/usr/local/bin # supervisord might live here\npackaging/monasca-agent.init.template:SUPERVISOR_PIDFILE\u003d\"/var/tmp/monasca-agent-supervisord.pid\"\npackaging/monasca-agent.init.template:SUPERVISOR_FILE\u003d\"{config_dir}/supervisor.conf\"\npackaging/monasca-agent.init.template:SUPERVISOR_SOCK\u003d\"/var/tmp/monasca-agent-supervisor.sock\"\npackaging/monasca-agent.init.template:SUPERVISORD\u003d\"{prefix}/bin/supervisord\"\npackaging/monasca-agent.init.template:SUPERVISORCTL\u003d\"{prefix}/bin/supervisorctl\"\npackaging/monasca-agent.init.template:    # If the socket exists, we can use supervisorctl\npackaging/monasca-agent.init.template:    if [ -e $SUPERVISOR_SOCK ]; then\npackaging/monasca-agent.init.template:        # If we\u0027re using supervisor, check the number of processes\npackaging/monasca-agent.init.template:        # supervisor is currently controlling, and make sure that it\u0027s the\npackaging/monasca-agent.init.template:        # same as the number of programs specified in the supervisor config\npackaging/monasca-agent.init.template:        supervisor_processes\u003d$($SUPERVISORCTL -c $SUPERVISOR_FILE status)\npackaging/monasca-agent.init.template:        supervisor_processes\u003d$(echo \"$supervisor_processes\" |\npackaging/monasca-agent.init.template:        supervisor_config_programs\u003d$(grep -v pup $SUPERVISOR_FILE |\npackaging/monasca-agent.init.template:        if [ \"$supervisor_processes\" -ne \"$supervisor_config_programs\" ]; then\npackaging/monasca-agent.init.template:            echo \"$supervisor_processes\"\npackaging/monasca-agent.init.template:            echo \"$DESC (supervisor) is NOT running all child processes\"\npackaging/monasca-agent.init.template:            echo \"$DESC (supervisor) is running all child processes\"\npackaging/monasca-agent.init.template:        echo \"$DESC (supervisor) is not running\"\npackaging/monasca-agent.init.template:        echo \"Starting $DESC (using supervisord)\" \"$NAME\"\npackaging/monasca-agent.init.template:        $SUPERVISORD -c $SUPERVISOR_FILE -u $AGENTUSER --pidfile $SUPERVISOR_PIDFILE\npackaging/monasca-agent.init.template:        if [ -e $SUPERVISOR_PIDFILE ]; then\npackaging/monasca-agent.init.template:            kill `cat $SUPERVISOR_PIDFILE`\npackaging/monasca-agent.init.template:            echo \"Pid file $SUPERVISOR_PIDFILE not found, nothing to stop\"\npackaging/monasca-agent/source/supervisord.conf:[supervisord]\npackaging/monasca-agent/source/supervisord.conf:logfile \u003d supervisord/logs/supervisord.log\npackaging/monasca-agent/source/supervisord.conf:identifier \u003d supervisor\npackaging/monasca-agent/source/supervisord.conf:pidfile \u003d supervisord/supervisord.pid\npackaging/monasca-agent/source/supervisord.conf:directory\u003d supervisord\npackaging/monasca-agent/source/supervisord.conf:[rpcinterface:supervisor]\npackaging/monasca-agent/source/supervisord.conf:supervisor.rpcinterface_factory \u003d supervisor.rpcinterface:make_main_rpcinterface\npackaging/monasca-agent/source/supervisord.conf:file \u003d supervisord/agent-supervisor.sock\npackaging/monasca-agent/source/supervisord.conf:[supervisorctl]\npackaging/monasca-agent/source/supervisord.conf:serverurl \u003d unix://supervisord/agent-supervisor.sock\npackaging/monasca-agent/source/supervisord.conf:stdout_logfile\u003dsupervisord/logs/collector.log\npackaging/monasca-agent/source/supervisord.conf:stdout_logfile\u003dsupervisord/logs/forwarder.log\npackaging/monasca-agent/source/supervisord.conf:stdout_logfile\u003dsupervisord/logs/dogstatsd.log\npackaging/monasca-agent/source/supervisord.conf:stdout_logfile\u003dsupervisord/logs/pup.log\npackaging/monasca-agent/source/setup_agent.sh:# set up supervisor\npackaging/monasca-agent/source/setup_agent.sh:printf \"Setting up supervisor.....\" | tee -a $logfile\npackaging/monasca-agent/source/setup_agent.sh:mkdir -p $dd_base/supervisord/logs \u003e\u003e $logfile 2\u003e\u00261\npackaging/monasca-agent/source/setup_agent.sh:$dd_base/venv/bin/pip install supervisor\u003d\u003d3.0b2 \u003e\u003e $logfile 2\u003e\u00261\npackaging/monasca-agent/source/setup_agent.sh:cp $dd_base/agent/packaging/datadog-agent/source/supervisord.conf $dd_base/supervisord/supervisord.conf \u003e\u003e $logfile 2\u003e\u00261\npackaging/monasca-agent/source/setup_agent.sh:ln -s $dd_base/supervisord/logs $dd_base/logs/supervisord \u003e\u003e $logfile 2\u003e\u00261\npackaging/monasca-agent/source/setup_agent.sh:    supervisord -c $dd_base/supervisord/supervisord.conf \u003e\u003e $logfile 2\u003e\u00261 \u0026\npackaging/monasca-agent/source/setup_agent.sh:    # Checking that supervisord was properly launched\npackaging/monasca-agent/source/setup_agent.sh:    supervisor_running\u003d$?\npackaging/monasca-agent/source/setup_agent.sh:    if [ $supervisor_running !\u003d 0 ]; then\npackaging/monasca-agent/source/setup_agent.sh:      unknown_error \"Failure when launching supervisor.\\n\"\npackaging/monasca-agent/source/agent:SUPERVISOR_NOT_RUNNING\u003d\"Supervisor is not running\"\npackaging/monasca-agent/source/agent:SUPERVISOR_CONF_FILE\u003d\u0027supervisord/supervisord.conf\u0027\npackaging/monasca-agent/source/agent:SOCK_FILE\u003d\u0027supervisord/agent-supervisor.sock\u0027\npackaging/monasca-agent/source/agent:supervisor_running() {\npackaging/monasca-agent/source/agent:    # Returns true if the supervisor is running, and false if not.\npackaging/monasca-agent/source/agent:    # We check if the supervisor is running by checking if\npackaging/monasca-agent/source/agent:execute_if_supervisor_running() {\npackaging/monasca-agent/source/agent:    # Executes `supervisord -c $SUPERVISOR_CONF_FILE $*`\npackaging/monasca-agent/source/agent:    # if the supervisor is running.\npackaging/monasca-agent/source/agent:    if supervisor_running; then\npackaging/monasca-agent/source/agent:        supervisorctl -c $SUPERVISOR_CONF_FILE \"$*\"\npackaging/monasca-agent/source/agent:        echo $SUPERVISOR_NOT_RUNNING\npackaging/monasca-agent/source/agent:    # First, we check if supervisor is running:\npackaging/monasca-agent/source/agent:    if ! supervisor_running; then\npackaging/monasca-agent/source/agent:        echo $SUPERVISOR_NOT_RUNNING\npackaging/monasca-agent/source/agent:        # Next, we check if supervisor is running all the datadog processes:\npackaging/monasca-agent/source/agent:        # Check the number of datadog processes supervisor is\npackaging/monasca-agent/source/agent:        # supervisor config file:\npackaging/monasca-agent/source/agent:        status_output\u003d$(supervisorctl -c $SUPERVISOR_CONF_FILE status)\npackaging/monasca-agent/source/agent:        datadog_supervisor_processes\u003d$(echo \"$status_output\" |\npackaging/monasca-agent/source/agent:        supervisor_config_programs\u003d$(grep -v pup $SUPERVISOR_CONF_FILE |\npackaging/monasca-agent/source/agent:        if [ \"$datadog_supervisor_processes\" -ne \"$supervisor_config_programs\" ]; then\npackaging/monasca-agent/source/agent:            echo \"Supervisor is NOT running all child processes\"\npackaging/monasca-agent/source/agent:            echo \"Supervisor is running all child processes\"\npackaging/monasca-agent/source/agent:        if supervisor_running; then\npackaging/monasca-agent/source/agent:            echo \"Supervisor is already running\"\npackaging/monasca-agent/source/agent:            echo \"Starting supervisor\"\npackaging/monasca-agent/source/agent:            supervisord -c $SUPERVISOR_CONF_FILE\npackaging/monasca-agent/source/agent:        execute_if_supervisor_running stop all\npackaging/monasca-agent/source/agent:        # FIXME: This command won\u0027t do anything if the supervisor isn\u0027t running.\npackaging/monasca-agent/source/agent:        # rely on the supervisor running. Once we stop starting the Agent in the\npackaging/monasca-agent/source/agent:        execute_if_supervisor_running restart all\npackaging/monasca-agent/smartos/dd-agent:\t\tcd $dd_base \u0026\u0026 $dd_base/venv/bin/supervisord -n -c $dd_base/supervisord/supervisord.conf\npackaging/monasca-agent/smartos/dd-agent:\t\tcd $dd_base \u0026\u0026 $dd_base/venv/bin/supervisorctl  -c $dd_base/supervisord/supervisord.conf shutdown\npackaging/monasca-agent-deb/supervisor.conf:[supervisorctl]\npackaging/monasca-agent-deb/supervisor.conf:serverurl \u003d unix:///var/tmp/monasca-agent-supervisor.sock\npackaging/monasca-agent-deb/supervisor.conf:file\u003d/var/tmp/monasca-agent-supervisor.sock\npackaging/monasca-agent-deb/supervisor.conf:[rpcinterface:supervisor]\npackaging/monasca-agent-deb/supervisor.conf:supervisor.rpcinterface_factory \u003d supervisor.rpcinterface:make_main_rpcinterface\npackaging/monasca-agent-deb/supervisor.conf:[supervisord]\npackaging/monasca-agent-deb/supervisor.conf:logfile \u003d /var/log/monasca/agent/supervisord.log\npackaging/monasca-agent-deb/supervisor.conf:pidfile \u003d /var/run/monasca-agent-supervisord.pid\npackaging/monasca-agent-deb/monasca-agent.init:SUPERVISOR_PIDFILE\u003d\"/var/run/monasca-agent-supervisord.pid\"\npackaging/monasca-agent-deb/monasca-agent.init:SUPERVISOR_FILE\u003d\"/etc/monasca/agent/supervisor.conf\"\npackaging/monasca-agent-deb/monasca-agent.init:SUPERVISOR_SOCK\u003d\"/var/tmp/monasca-agent-supervisor.sock\"\npackaging/monasca-agent-deb/monasca-agent.init:SUPERVISORD\u003d$(which supervisord)\npackaging/monasca-agent-deb/monasca-agent.init:    # If the socket exists, we can use supervisorctl\npackaging/monasca-agent-deb/monasca-agent.init:    if [ -e $SUPERVISOR_SOCK ]; then\npackaging/monasca-agent-deb/monasca-agent.init:        # If we\u0027re using supervisor, check the number of processes\npackaging/monasca-agent-deb/monasca-agent.init:        # supervisor is currently controlling, and make sure that it\u0027s the\npackaging/monasca-agent-deb/monasca-agent.init:        # same as the number of programs specified in the supervisor config\npackaging/monasca-agent-deb/monasca-agent.init:        supervisor_processes\u003d$(supervisorctl -c $SUPERVISOR_FILE status)\npackaging/monasca-agent-deb/monasca-agent.init:        supervisor_processes\u003d$(echo \"$supervisor_processes\" |\npackaging/monasca-agent-deb/monasca-agent.init:        supervisor_config_programs\u003d$(grep -v pup $SUPERVISOR_FILE |\npackaging/monasca-agent-deb/monasca-agent.init:        if [ \"$supervisor_processes\" -ne \"$supervisor_config_programs\" ]; then\npackaging/monasca-agent-deb/monasca-agent.init:            echo \"$supervisor_processes\"\npackaging/monasca-agent-deb/monasca-agent.init:            echo \"$DESC (supervisor) is NOT running all child processes\"\npackaging/monasca-agent-deb/monasca-agent.init:            echo \"$DESC (supervisor) is running all child processes\"\npackaging/monasca-agent-deb/monasca-agent.init:        echo \"$DESC (supervisor) is not running\"\npackaging/monasca-agent-deb/monasca-agent.init:        log_daemon_msg \"Starting $DESC (using supervisord)\" \"$NAME\"\npackaging/monasca-agent-deb/monasca-agent.init:        start-stop-daemon --start --quiet --oknodo --exec $SUPERVISORD -- -c $SUPERVISOR_FILE --pidfile $SUPERVISOR_PIDFILE\npackaging/monasca-agent-deb/monasca-agent.init:        log_daemon_msg \"Stopping $DESC (stopping supervisord)\" \"$NAME\"\npackaging/monasca-agent-deb/monasca-agent.init:        start-stop-daemon --stop --retry 30 --quiet --oknodo --pidfile $SUPERVISOR_PIDFILE\npackaging/Makefile:# Make the monasca agent debian package that includes supervisor, the forwarder\n```","accounts_in_message":[],"_revision_number":26},{"id":"f1dd67c678dc38752214d659555599a43c20c5bc","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-06-28 07:54:17.000000000","message":"Patch Set 26:\n\nDoug, I think we should not handle packaging in this change.\n\nThanks for pointing to this though. It hasn\u0027t been maintained for a longer time but I see some value in having universal packaging tooling. Let\u0027s discuss it separately and decide how to proceed.","accounts_in_message":[],"_revision_number":26},{"id":"28a3234725b5761ba04734ae806adb9806b9423f","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-28 09:06:33.000000000","message":"Patch Set 26:\n\nI agree with Witek","accounts_in_message":[],"_revision_number":26},{"id":"9c9a9123711322d3bb8d3f3084b82021253bca16","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-06-28 09:43:23.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\nOk, sounds good. I just tested it out and it works for me. Thanks!\n\n```\n(inst) [doug@dev system]$ systemctl status monasca*\n● monasca-agent.target - Monasca Agent\n   Loaded: loaded (/etc/systemd/system/monasca-agent.target; enabled; vendor preset: disabled)\n   Active: active since Thu 2018-06-28 10:39:35 BST; 1min 44s ago\n\n● monasca-collector.service - Monasca Agent - Collector\n   Loaded: loaded (/etc/systemd/system/monasca-collector.service; disabled; vendor preset: disabled)\n   Active: active (running) since Thu 2018-06-28 10:39:35 BST; 1min 44s ago\n Main PID: 115941 (monasca-collect)\n    Tasks: 6\n   CGroup: /system.slice/monasca-collector.service\n           └─115941 /opt/doug/git/monasca-agent/inst/bin/python2 /opt/doug/git/monasca-agent/inst/bin/monasca-collector foreground\n\n● monasca-forwarder.service - Monasca Agent - Forwarder\n   Loaded: loaded (/etc/systemd/system/monasca-forwarder.service; disabled; vendor preset: disabled)\n   Active: active (running) since Thu 2018-06-28 10:39:35 BST; 1min 44s ago\n Main PID: 115939 (monasca-forward)\n    Tasks: 1\n   CGroup: /system.slice/monasca-forwarder.service\n           └─115939 /opt/doug/git/monasca-agent/inst/bin/python2 /opt/doug/git/monasca-agent/inst/bin/monasca-forwarder\n\n● monasca-statsd.service - Monasca Agent - Statsd\n   Loaded: loaded (/etc/systemd/system/monasca-statsd.service; disabled; vendor preset: disabled)\n   Active: active (running) since Thu 2018-06-28 10:39:35 BST; 1min 44s ago\n Main PID: 115940 (monasca-statsd)\n    Tasks: 2\n   CGroup: /system.slice/monasca-statsd.service\n           └─115940 /opt/doug/git/monasca-agent/inst/bin/python2 /opt/doug/git/monasca-agent/inst/bin/monasca-statsd\n```","accounts_in_message":[],"_revision_number":26},{"id":"671d6ac87c6a388377bc8e95fe60ca9aefea0882","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-28 12:02:46.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"9e6caa01138bebc4003519cdfd55ab662100b7a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 13:00:37.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/26/check/openstack-tox-pep8/d941783/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/26/check/openstack-tox-py27/ae9f1fd/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/26/check/openstack-tox-py35/01baf3f/ : SUCCESS in 6m 37s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-python-influxdb/6e17d89/ : SUCCESS in 43m 19s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-java-influxdb/de87bb3/ : SUCCESS in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"cabd4e8563761939504446fde471fbebd6c3fbbd","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2018-06-29 12:50:53.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"6797d795309d9299f4975d3c38b32b74204740c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 13:49:15.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/26/check/openstack-tox-pep8/ec5ccc1/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/26/check/openstack-tox-py27/b5baff1/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/04/554304/26/check/openstack-tox-py35/b16c180/ : SUCCESS in 5m 28s (non-voting)\n- monasca-tempest-python-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-python-influxdb/c5eda0d/ : SUCCESS in 48m 01s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/04/554304/26/check/monasca-tempest-java-influxdb/d1d94b8/ : SUCCESS in 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c4e0aed0bf59ab8c1aaf304a9791cb400dde76c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 13:57:47.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"cc9266445ab15a54b4457a067d7785b111890bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 14:20:41.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/04/554304/26/gate/openstack-tox-pep8/892e6ed/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/04/554304/26/gate/openstack-tox-py27/f36b8c5/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":26},{"id":"d2a2e2748c27ab7fe3db1efefebc16a53f3d863f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 14:20:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"a8a2bb845b1129d24683d462c5d1ab08955a2840","revisions":{"1377d83663bb97debaa1ec6c19e6d445c21bda82":{"kind":"REWORK","_number":1,"created":"2018-03-19 17:40:43.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/1"}}},"commit":{"parents":[{"commit":"ceca989b2abfcc65c6f9f1ab139eb49d782eabc9","subject":"Merge \"Add LXC swap metric collector and fix lxc bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ceca989b2abfcc65c6f9f1ab139eb49d782eabc9"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:40:41.000000000","tz":0},"subject":"First step towards python 3 support.","message":"First step towards python 3 support.\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1377d83663bb97debaa1ec6c19e6d445c21bda82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1377d83663bb97debaa1ec6c19e6d445c21bda82"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ceca989b2abfcc65c6f9f1ab139eb49d782eabc9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ff2dd0b00e9a22fb6980d3e18d4efba313497c10":{"kind":"REWORK","_number":2,"created":"2018-03-19 21:32:33.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/2"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 21:32:15.000000000","tz":0},"subject":"First step towards python 3 support.","message":"First step towards python 3 support.\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ff2dd0b00e9a22fb6980d3e18d4efba313497c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ff2dd0b00e9a22fb6980d3e18d4efba313497c10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4424f9fd37227f203f4daf40641578e37178c58d":{"kind":"REWORK","_number":3,"created":"2018-03-22 17:22:07.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/3"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-22 17:21:05.000000000","tz":0},"subject":"First step towards python 3 support.","message":"First step towards python 3 support.\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4424f9fd37227f203f4daf40641578e37178c58d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4424f9fd37227f203f4daf40641578e37178c58d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"49d249d78fa8e44595a2d6889c114f00c47903d6":{"kind":"NO_CHANGE","_number":4,"created":"2018-03-23 22:14:39.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/4"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-23 22:14:08.000000000","tz":0},"subject":"First step towards python 3 support.","message":"First step towards python 3 support.\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/49d249d78fa8e44595a2d6889c114f00c47903d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/49d249d78fa8e44595a2d6889c114f00c47903d6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2c628d5f62253e00dcd5c5298598adfb7495827b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2018-03-23 22:17:09.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/5"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-23 22:16:55.000000000","tz":0},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2c628d5f62253e00dcd5c5298598adfb7495827b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2c628d5f62253e00dcd5c5298598adfb7495827b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b9a6c2e2908bbc4dc2d16ecc1bb8c8ca26af5221":{"kind":"REWORK","_number":6,"created":"2018-03-28 20:23:18.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/6"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-28 20:22:48.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\nRemoved dependacy on supervisor.\nAdded template configuration for systemd target that includes all\nservices.\nAdded templates configuration for systemd service for every single\nservice.\nChanged monasca_setup to use the new templates.\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\nTask: 4126\nStory: 2000975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b9a6c2e2908bbc4dc2d16ecc1bb8c8ca26af5221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b9a6c2e2908bbc4dc2d16ecc1bb8c8ca26af5221"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"61e1e8fa4edeb3fc09c2c108297ab368511ffa3d":{"kind":"REWORK","_number":7,"created":"2018-04-10 20:56:17.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/7"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-04-10 20:55:46.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/61e1e8fa4edeb3fc09c2c108297ab368511ffa3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/61e1e8fa4edeb3fc09c2c108297ab368511ffa3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ca918eafdebdf0dd7f941c681a0aedca23de609f":{"kind":"REWORK","_number":8,"created":"2018-04-11 20:34:42.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/8"}}},"commit":{"parents":[{"commit":"332fe52879eb7a217409566dd398600dc0ab51eb","subject":"Merge \"Update supervisor version lower bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/332fe52879eb7a217409566dd398600dc0ab51eb"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-04-11 20:34:19.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ca918eafdebdf0dd7f941c681a0aedca23de609f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ca918eafdebdf0dd7f941c681a0aedca23de609f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"332fe52879eb7a217409566dd398600dc0ab51eb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d5d5e1f9c44ae30b9d5b6c6a1426842cb23b47e0":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-04-11 21:24:41.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/9"}}},"commit":{"parents":[{"commit":"cdff90a106270acc42e20a64caddb0c036f4fe2e","subject":"Merge \"Add support for k8s StatefulSet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/cdff90a106270acc42e20a64caddb0c036f4fe2e"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa (sc)","email":"sc@linux.it","date":"2018-04-11 21:24:41.000000000","tz":0},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d5d5e1f9c44ae30b9d5b6c6a1426842cb23b47e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d5d5e1f9c44ae30b9d5b6c6a1426842cb23b47e0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cdff90a106270acc42e20a64caddb0c036f4fe2e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"61a0ae6cab5c17fc95775e31b2e51abfdf655b84":{"kind":"REWORK","_number":10,"created":"2018-05-02 15:10:16.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/10"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-02 08:25:56.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/61a0ae6cab5c17fc95775e31b2e51abfdf655b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/61a0ae6cab5c17fc95775e31b2e51abfdf655b84"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"89ab8d41f177b2b72260f2a5c2da7a67b7db1ce0":{"kind":"REWORK","_number":11,"created":"2018-05-03 01:06:09.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/11"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-03 01:05:53.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/89ab8d41f177b2b72260f2a5c2da7a67b7db1ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/89ab8d41f177b2b72260f2a5c2da7a67b7db1ce0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b99275147bf1d01586a4c983277eda46ecef28e5":{"kind":"REWORK","_number":12,"created":"2018-05-04 22:36:47.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/12"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-04 22:36:02.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b99275147bf1d01586a4c983277eda46ecef28e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b99275147bf1d01586a4c983277eda46ecef28e5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4e1caad8f06bbc105d4df690cf6cd07ac8a23f10":{"kind":"REWORK","_number":13,"created":"2018-05-06 20:46:48.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/13"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-06 20:44:49.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4e1caad8f06bbc105d4df690cf6cd07ac8a23f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4e1caad8f06bbc105d4df690cf6cd07ac8a23f10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0f252bde807b196beb819c91ea7fb7706f66bdef":{"kind":"REWORK","_number":14,"created":"2018-05-06 21:06:05.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/14"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-06 21:05:53.000000000","tz":60},"subject":"First step towards python 3 support","message":"First step towards python 3 support\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0f252bde807b196beb819c91ea7fb7706f66bdef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0f252bde807b196beb819c91ea7fb7706f66bdef"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e5972a403b8756ce880126fed01eb4ba6704ade8":{"kind":"REWORK","_number":15,"created":"2018-05-07 14:43:41.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/15"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-07 14:38:21.000000000","tz":60},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to 80 columns and some other small\nchanges were done to comply with pycodestyle and pydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/e5972a403b8756ce880126fed01eb4ba6704ade8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/e5972a403b8756ce880126fed01eb4ba6704ade8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a18d4e01eb6ebb16029cbe2db0b1c00ace14b241":{"kind":"REWORK","_number":16,"created":"2018-05-07 15:28:25.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/16"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-07 15:27:18.000000000","tz":60},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a18d4e01eb6ebb16029cbe2db0b1c00ace14b241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a18d4e01eb6ebb16029cbe2db0b1c00ace14b241"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"657ba0a83d5d2c854a6c4d476e82626e2cdf5347":{"kind":"REWORK","_number":17,"created":"2018-05-07 19:35:28.000000000","uploader":{"_account_id":14123,"name":"Stefano Canepa","email":"sc@linux.it","username":"sc"},"ref":"refs/changes/04/554304/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/17"}}},"commit":{"parents":[{"commit":"2b2b5719c75076132b251aa811796feffed82880","subject":"Fix type error in monasca agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2b2b5719c75076132b251aa811796feffed82880"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-05-07 19:35:04.000000000","tz":60},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/657ba0a83d5d2c854a6c4d476e82626e2cdf5347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/657ba0a83d5d2c854a6c4d476e82626e2cdf5347"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b2b5719c75076132b251aa811796feffed82880","is_merged_in_target_branch":true,"change_id":"Ibaf148ca6cf7b9cb73ceabacccf7aa519897af72","change_number":564501,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"42d679077a8d0b503d3bfa22312161512c851b90":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2018-06-13 09:04:52.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/18"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-13 09:02:04.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/42d679077a8d0b503d3bfa22312161512c851b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/42d679077a8d0b503d3bfa22312161512c851b90"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"19720d3a944a800bcfe8a65f88065478da8655ea":{"kind":"REWORK","_number":19,"created":"2018-06-25 08:33:50.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/19"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-25 08:33:36.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/19720d3a944a800bcfe8a65f88065478da8655ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/19720d3a944a800bcfe8a65f88065478da8655ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"02dc1a8746cc3566416af3ef394645dfd3fbe96b":{"kind":"REWORK","_number":20,"created":"2018-06-25 08:35:52.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/20"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-25 08:35:41.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/02dc1a8746cc3566416af3ef394645dfd3fbe96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/02dc1a8746cc3566416af3ef394645dfd3fbe96b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0f483ed4db23433c621bd404b96dbf0a764965bc":{"kind":"REWORK","_number":21,"created":"2018-06-25 08:43:51.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/21"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-25 08:43:31.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0f483ed4db23433c621bd404b96dbf0a764965bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0f483ed4db23433c621bd404b96dbf0a764965bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8e8b76f88387acef14b0474f300a549025e7b0f8":{"kind":"REWORK","_number":22,"created":"2018-06-25 09:11:43.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/22"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-25 09:11:34.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8e8b76f88387acef14b0474f300a549025e7b0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8e8b76f88387acef14b0474f300a549025e7b0f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"657a4e48ae67b2b0a581bd5513d446d381a0d7ad":{"kind":"REWORK","_number":23,"created":"2018-06-25 13:32:55.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/23"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-25 13:32:46.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/657a4e48ae67b2b0a581bd5513d446d381a0d7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/657a4e48ae67b2b0a581bd5513d446d381a0d7ad"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0ae129f6e585cff0bd93001fefd3e682337adb29":{"kind":"REWORK","_number":24,"created":"2018-06-26 06:42:45.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/24"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-26 06:42:31.000000000","tz":120},"subject":"Removed dependacy on supervisor","message":"Removed dependacy on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependacy on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0ae129f6e585cff0bd93001fefd3e682337adb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/0ae129f6e585cff0bd93001fefd3e682337adb29"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"915acdddbc7a10f7878c73fb4896e6db098bb66b":{"kind":"REWORK","_number":25,"created":"2018-06-27 08:47:44.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/25"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-27 08:46:51.000000000","tz":120},"subject":"Removed dependency on supervisor","message":"Removed dependency on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependency on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/915acdddbc7a10f7878c73fb4896e6db098bb66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/915acdddbc7a10f7878c73fb4896e6db098bb66b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a8a2bb845b1129d24683d462c5d1ab08955a2840":{"kind":"REWORK","_number":26,"created":"2018-06-27 09:28:53.000000000","uploader":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"ref":"refs/changes/04/554304/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/04/554304/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/04/554304/26"}}},"commit":{"parents":[{"commit":"dd66bd150421ccc80d9469cfc253a94fc9e56041","subject":"Merge \"fix misspelling of \u0027monasca\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dd66bd150421ccc80d9469cfc253a94fc9e56041"}]}],"author":{"name":"Stefano Canepa","email":"sc@linux.it","date":"2018-03-19 17:14:47.000000000","tz":0},"committer":{"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","date":"2018-06-27 09:28:43.000000000","tz":120},"subject":"Removed dependency on supervisor","message":"Removed dependency on supervisor\n\nTo support python3 in the near future this was done:\n\n* Removed dependency on supervisor.\n* Added template configuration for systemd target that includes all\n  services.\n* Added templates configuration for systemd service for every single\n  service.\n* Changed monasca_setup to use the new templates.\n\nIn the meanwhile code was formated to cope with pep8 settings and some\nother small changes were done to comply with pycodestyle and\npydocstring.\n\nTask: 4126\nStory: 2000975\n\nDepends-On: https://review.openstack.org/#/c/566475/\n\nChange-Id: I0d0c4ea41a830581d6b9f247fad6a2dda1f96cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a8a2bb845b1129d24683d462c5d1ab08955a2840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a8a2bb845b1129d24683d462c5d1ab08955a2840"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dd66bd150421ccc80d9469cfc253a94fc9e56041","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
