)]}'
{"id":"openstack%2Fmonasca-agent~623611","triplet_id":"openstack%2Fmonasca-agent~master~I8050e1eed68d7b64f7a968b061afa69fe2e86d72","project":"openstack/monasca-agent","branch":"master","topic":"story2004539-remove-matching","hashtags":[],"change_id":"I8050e1eed68d7b64f7a968b061afa69fe2e86d72","subject":"Add remove configuration for matching arguments","status":"MERGED","created":"2018-12-08 01:12:13.000000000","updated":"2020-07-20 16:28:16.000000000","submitted":"2020-04-24 10:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"623611-1587725732424-602cea05","meta_rev_id":"2e0b8583e05e2167cf3ae9ae6780bcf3fcd6aaf7","_number":623611,"virtual_id_number":623611,"owner":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 10:55:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":2,"date":"2020-04-24 10:43:42.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":1,"date":"2020-04-24 10:43:42.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-10 00:34:54.000000000","updated_by":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"reviewer":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"state":"REVIEWER"},{"updated":"2020-03-03 15:02:28.000000000","updated_by":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2020-04-24 10:43:42.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-04-24 10:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f35eba4125e16e844e46e03f08dd351513573706","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-12-08 01:12:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"557ffb4ef1b503c7862e158a2478875b319c6e09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 02:02:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/623611/1/check/build-openstack-sphinx-docs/ee5724f/ : POST_FAILURE in 5m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/11/623611/1/check/openstack-tox-pep8/93d387d/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/11/623611/1/check/openstack-tox-py27/8397ba0/ : SUCCESS in 5m 39s\n- openstack-tox-py35 http://logs.openstack.org/11/623611/1/check/openstack-tox-py35/ef50c55/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/11/623611/1/check/openstack-tox-py36/0173ed4/ : SUCCESS in 5m 05s\n- monasca-tempest-python-influxdb http://logs.openstack.org/11/623611/1/check/monasca-tempest-python-influxdb/a78ecd4/ : SUCCESS in 30m 56s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/11/623611/1/check/monasca-tempest-java-influxdb/3efb5f0/ : FAILURE in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0552fe316b2da02f15d9ef41f58e29d14ffced35","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-04-04 22:47:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54ab21df552344e8fb2619a0846e90a11bcb29fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 23:41:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/11/623611/2/check/build-openstack-sphinx-docs/a7d9a84/ : POST_FAILURE in 6m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/11/623611/2/check/openstack-tox-pep8/46431dc/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/11/623611/2/check/openstack-tox-py27/460a434/ : FAILURE in 8m 03s\n- openstack-tox-py35 http://logs.openstack.org/11/623611/2/check/openstack-tox-py35/7f2e4d4/ : SUCCESS in 6m 20s\n- openstack-tox-py36 http://logs.openstack.org/11/623611/2/check/openstack-tox-py36/3406f37/ : SUCCESS in 7m 13s\n- openstack-tox-py37 http://logs.openstack.org/11/623611/2/check/openstack-tox-py37/5779a21/ : SUCCESS in 7m 09s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/11/623611/2/check/monasca-tempest-python2-influxdb/78562e1/ : FAILURE in 44m 08s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/11/623611/2/check/monasca-tempest-python3-influxdb/24bfb18/ : SUCCESS in 44m 17s (non-voting)\n- monasca-tempest-java-influxdb http://logs.openstack.org/11/623611/2/check/monasca-tempest-java-influxdb/29a7c1f/ : FAILURE in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3717845f0febd3586ab27d52cc9a5bb4ea669db","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-08-01 22:42:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2f0a979fc2fd5e20c18e4219422a32c9e9c5fed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 23:37:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://logs.opendev.org/11/623611/3/check/openstack-tox-pep8/99bcaac/ : SUCCESS in 4m 59s\n- openstack-tox-py27 https://logs.opendev.org/11/623611/3/check/openstack-tox-py27/b737ec3/ : FAILURE in 7m 31s\n- openstack-tox-py36 https://logs.opendev.org/11/623611/3/check/openstack-tox-py36/efed0b8/ : FAILURE in 7m 25s\n- openstack-tox-py37 https://logs.opendev.org/11/623611/3/check/openstack-tox-py37/05fb0aa/ : SUCCESS in 6m 56s\n- build-monasca-docker-image https://logs.opendev.org/11/623611/3/check/build-monasca-docker-image/e1fc804/ : SUCCESS in 14m 21s\n- monasca-tempest-python2-influxdb https://logs.opendev.org/11/623611/3/check/monasca-tempest-python2-influxdb/8a9006c/ : SUCCESS in 39m 06s (non-voting)\n- monasca-tempest-python3-influxdb https://logs.opendev.org/11/623611/3/check/monasca-tempest-python3-influxdb/334983e/ : SUCCESS in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e553b23c682655fda758cef7ebd41a4ba6c234b","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-09-09 21:40:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4e0b860f146a98aa9e0ad68a0bb427f907285f6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 22:42:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25934e8a95c481fb0df3c3d431216af : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b925b1e9354a40b6bd793481f52738b1 : FAILURE in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f2c0f1d3fef48a995bdfd338a405d00 : FAILURE in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/412def6ca493403ba8bd65930addc2eb : SUCCESS in 8m 47s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/cb694f079694432fb1a24b15a4b4eb36 : SUCCESS in 19m 32s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/6d7c61fca8aa47f88c11a8f768e99535 : SUCCESS in 55m 38s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/7fafa41ced2f4931a0964fa7b11cb29b : FAILURE in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e749a0c5f6ce07a7226feb4d9f3c1b7ff93a383","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-02-20 05:54:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f84a7a66bf7207a439d2a99993d439aed5c0a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 06:45:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c651a34624f640beb4662810a4bf6379 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834b8d83d8b24a86a3cca4a03a43b0ee : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9f7929e644d4eb790d43711bc06ff33 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd7f7d9ee8944c893c20a988214ed29 : SUCCESS in 8m 56s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/2ee6af38227543d1be632c89194e2efd : SUCCESS in 27m 07s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/317eba87e6cf47f29d1fda7ace056b52 : FAILURE in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9d20f249168b263051ecd85895e08217bb0b401","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-04-23 17:46:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"895b14baacbeb87c17e3c6a55349f5191ec8f420","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-04-23 17:49:03.000000000","message":"Patch Set 5:\n\nCould we add argument documentation in https://opendev.org/openstack/monasca-agent/src/branch/master/docs/Agent.md#user-content-explanation-of-primary-monasca-setup-command-line-parameters ? Could be done as a follow-up change.\n\nThe DeveloperDocs are very difficult to find.","accounts_in_message":[],"_revision_number":5},{"id":"1e71e61ee29afbd02f2cbbb001291dc87575788d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 18:25:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62ebd7966ea4d369adba304e67f81d3 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/176f78b7f57043d0a6bf8a3e1dc80aec : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/085d250404ac4228b3f602d0f12587e0 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cf6bbbebc7a452796be6db7842ae0ac : SUCCESS in 5m 13s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/3c636e71a29a47d887e533942352f341 : SUCCESS in 26m 48s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/04686e24140b4f56aeb9ce7457eb4924 : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43cc5aa5302dabd9f2c39ff27748cb773a5651b8","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-23 18:57:27.000000000","message":"Patch Set 5: Workflow-1\n\n\u003e Could we add argument documentation in https://opendev.org/openstack/monasca-agent/src/branch/master/docs/Agent.md#user-content-explanation-of-primary-monasca-setup-command-line-parameters\n \u003e ? Could be done as a follow-up change.\n \u003e \n \u003e The DeveloperDocs are very difficult to find.\n\nGood idea.  I will see if I can add something today, otherwise we can do it as a follow on commit.","accounts_in_message":[],"_revision_number":5},{"id":"c5825f8b8567e3fbabf92fb0b9bf66637a6068b0","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-23 22:47:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"878bab3503dc74d7bd4febd5830eadbc08b0a2b8","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-23 22:49:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e177bed9fb31ea935a6bf0c176618547730f6681","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-23 22:56:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2baae921c9ea8533980e6b27cff9e619620cea16","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-23 22:57:28.000000000","message":"Patch Set 8: Review-Priority+1\n\nThanks again for the suggestion. I think this is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"584efcdf4d270fb17cb8dc33003f23ded8b1b56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 23:41:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aab29495db74e2f94ebe401624c9df7 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cb7b3abf15a4167a9f52d1ebe115248 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dc4bde2a9234b1cbbf8eb05b128e529 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e8cc9d923343c9bdd155d23a54ab45 : SUCCESS in 5m 11s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/414c8f1024d741bc81e05d6f74668805 : SUCCESS in 27m 10s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/387fb10889fc48f392a7069d9d6f2fe9 : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e0692b4fe668ad096d22868e48e2069117e5da0e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-04-24 10:43:42.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":8},{"id":"8af06e845bb780bd2cd2806a75a00be700ec9a07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:43:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"246adaad3629d1cc4b412d02fd8b39c0ff9158c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:55:32.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cc98f853134671be926611fee654d3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e1de8a95d24de6bb57a348dc3506b9 : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c911d1208b04863a75e2cceb6c50792 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":8},{"id":"a74efcd96d40e07686c495d10cfa9821f3a81997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:55:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"7454c422a736830451ae07d3a87ddd2546e1e999","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-07-20 16:28:16.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 3d914cc7029d4fa2ba1e56267857c0cb59c7cf9e","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8d4bd979d5789689ff5fe5298171e60ac0d65c81","revisions":{"71eb0aa56b356aa57265b3e836c9c088be73edea":{"kind":"REWORK","_number":1,"created":"2018-12-08 01:12:13.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/1"}}},"commit":{"parents":[{"commit":"7c3cce78f4515b16836b270e17a497a27146aeeb","subject":"Change openstack-dev to openstack-discuss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c3cce78f4515b16836b270e17a497a27146aeeb"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"subject":"WIP Add remove configuration for matching arguments","message":"WIP Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nWIP: this new function and the existing remove_config() need to\nhave unit tests added.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/71eb0aa56b356aa57265b3e836c9c088be73edea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/71eb0aa56b356aa57265b3e836c9c088be73edea"}]},"branch":"refs/heads/master"},"ff3aae3c679fe0411e4ca833049529f6ecafd2b1":{"kind":"REWORK","_number":2,"created":"2019-04-04 22:47:10.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/2"}}},"commit":{"parents":[{"commit":"7c3cce78f4515b16836b270e17a497a27146aeeb","subject":"Change openstack-dev to openstack-discuss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c3cce78f4515b16836b270e17a497a27146aeeb"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-04-04 22:46:58.000000000","tz":-420},"subject":"WIP Add remove configuration for matching arguments","message":"WIP Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nWIP: this new function and the existing remove_config() need to\nhave unit tests added.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ff3aae3c679fe0411e4ca833049529f6ecafd2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ff3aae3c679fe0411e4ca833049529f6ecafd2b1"}]},"branch":"refs/heads/master"},"9eacb72a27d386ea0ec6ec82210dd23a9e6ba38b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-01 22:42:53.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/3"}}},"commit":{"parents":[{"commit":"ea9f2ebc15c35a72f57c614a4b0501a08b8e66eb","subject":"Update api-ref location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ea9f2ebc15c35a72f57c614a4b0501a08b8e66eb"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-08-01 22:42:53.000000000","tz":0},"subject":"WIP Add remove configuration for matching arguments","message":"WIP Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nWIP: this new function and the existing remove_config() need to\nhave unit tests added.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/9eacb72a27d386ea0ec6ec82210dd23a9e6ba38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/9eacb72a27d386ea0ec6ec82210dd23a9e6ba38b"}]},"branch":"refs/heads/master"},"a72c788166a789ce6306a7da4bd4f972556d50ff":{"kind":"REWORK","_number":4,"created":"2020-02-20 05:54:35.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/4"}}},"commit":{"parents":[{"commit":"1caaf4423a9a779f56271df7db445289a76f7728","subject":"Close db connection in mysql check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1caaf4423a9a779f56271df7db445289a76f7728"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2020-02-20 05:54:22.000000000","tz":-480},"subject":"Add remove configuration for matching arguments","message":"Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a72c788166a789ce6306a7da4bd4f972556d50ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a72c788166a789ce6306a7da4bd4f972556d50ff"}]},"branch":"refs/heads/master"},"d8f87d6cd23d97f55ffda36dac3deab972508a35":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-23 17:46:23.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/11/623611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/5"}}},"commit":{"parents":[{"commit":"a69da1c28d907a734c57bebc314b220f82bd781e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a69da1c28d907a734c57bebc314b220f82bd781e"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Witold Bedyk","email":"witold.bedyk@suse.com","date":"2020-04-23 17:46:23.000000000","tz":0},"subject":"Add remove configuration for matching arguments","message":"Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d8f87d6cd23d97f55ffda36dac3deab972508a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d8f87d6cd23d97f55ffda36dac3deab972508a35"}]},"branch":"refs/heads/master"},"87a1343987135d0ad49ea718f453c73876f1d0aa":{"kind":"REWORK","_number":6,"created":"2020-04-23 22:47:49.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/6"}}},"commit":{"parents":[{"commit":"a69da1c28d907a734c57bebc314b220f82bd781e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a69da1c28d907a734c57bebc314b220f82bd781e"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2020-04-23 22:47:29.000000000","tz":-420},"subject":"Add remove configuration for matching arguments","message":"Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/87a1343987135d0ad49ea718f453c73876f1d0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/87a1343987135d0ad49ea718f453c73876f1d0aa"}]},"branch":"refs/heads/master"},"951238397d0b5a8bb30178f09eb70dd0b6d81b45":{"kind":"REWORK","_number":7,"created":"2020-04-23 22:49:19.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/7"}}},"commit":{"parents":[{"commit":"a69da1c28d907a734c57bebc314b220f82bd781e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a69da1c28d907a734c57bebc314b220f82bd781e"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2020-04-23 22:49:09.000000000","tz":-420},"subject":"Add remove configuration for matching arguments","message":"Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/951238397d0b5a8bb30178f09eb70dd0b6d81b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/951238397d0b5a8bb30178f09eb70dd0b6d81b45"}]},"branch":"refs/heads/master"},"8d4bd979d5789689ff5fe5298171e60ac0d65c81":{"kind":"REWORK","_number":8,"created":"2020-04-23 22:56:31.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/11/623611/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/11/623611/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/11/623611/8"}}},"commit":{"parents":[{"commit":"a69da1c28d907a734c57bebc314b220f82bd781e","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a69da1c28d907a734c57bebc314b220f82bd781e"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2018-12-08 01:07:42.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2020-04-23 22:56:21.000000000","tz":-420},"subject":"Add remove configuration for matching arguments","message":"Add remove configuration for matching arguments\n\nThe remove_config() function only removes an exact match for\nthe configured instance.\nThis will allow removing plugin configuration when all config is\nnot known.\n\nUse case: A compute node has been removed, so any host_alive\nping checks that are configured for it should be removed.  But at\nthe time of removal the list of target_hostnames to match are\nnot known.\n\nChange-Id: I8050e1eed68d7b64f7a968b061afa69fe2e86d72\nStory: 2004539\nTask: 28287\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8d4bd979d5789689ff5fe5298171e60ac0d65c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8d4bd979d5789689ff5fe5298171e60ac0d65c81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
