)]}'
{"id":"openstack%2Fmonasca-agent~672108","triplet_id":"openstack%2Fmonasca-agent~master~I1ea64ce2217776658e8139aa041e4e217985dfbb","project":"openstack/monasca-agent","branch":"master","topic":"fix-log","hashtags":[],"change_id":"I1ea64ce2217776658e8139aa041e4e217985dfbb","subject":"Add missing space to log statement","status":"MERGED","created":"2019-07-22 16:55:23.000000000","updated":"2019-07-23 08:12:41.000000000","submitted":"2019-07-23 08:12:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672108-1563869561074-a8feb72e","meta_rev_id":"452bbd6f718f6506eac264d88f655cfdb05b53bc","_number":672108,"virtual_id_number":672108,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-07-23 08:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-22 16:58:40.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-07-22 23:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-23 08:01:02.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 23:25:55.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-07-23 08:01:02.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-23 08:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a83fa59ed053ccb9ffd32abcb238211954e65ae5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-22 16:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54ee0bb54645f08ae8b0e340570e285b2e76ff7c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-22 16:58:40.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, I\u0027ve seen this too.","accounts_in_message":[],"_revision_number":1},{"id":"ac0086fd016be06540922c80154b88718b4624ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:09:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/672108/1/check/openstack-tox-pep8/d4c4275/ : SUCCESS in 7m 50s\n- openstack-tox-py27 http://logs.openstack.org/08/672108/1/check/openstack-tox-py27/efe19ba/ : SUCCESS in 7m 32s\n- openstack-tox-py36 http://logs.openstack.org/08/672108/1/check/openstack-tox-py36/2b14806/ : SUCCESS in 6m 38s\n- openstack-tox-py37 http://logs.openstack.org/08/672108/1/check/openstack-tox-py37/91acfe4/ : SUCCESS in 8m 39s\n- build-monasca-docker-image http://logs.openstack.org/08/672108/1/check/build-monasca-docker-image/c6589a8/ : SUCCESS in 15m 08s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/08/672108/1/check/monasca-tempest-python2-influxdb/14df1c7/ : SUCCESS in 42m 09s (non-voting)\n- monasca-tempest-python3-influxdb http://logs.openstack.org/08/672108/1/check/monasca-tempest-python3-influxdb/2f136e2/ : SUCCESS in 37m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"397da455e04291f65595e4a52c8579634fd9291e","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-07-22 23:25:55.000000000","message":"Patch Set 1: Code-Review+1\n\nlooks fine (and will read better)","accounts_in_message":[],"_revision_number":1},{"id":"8146112f873d34a40efcba5734563854ddf26b5c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-23 08:01:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5de4e3cf0ed595d295b268827c710170d252cb39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6df4650bd4d8c848092e34eb63bf1b7923300d0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:12:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/08/672108/1/gate/openstack-tox-pep8/aad9d1a/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/08/672108/1/gate/openstack-tox-py27/66fae1b/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/08/672108/1/gate/openstack-tox-py36/9a4b454/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/08/672108/1/gate/openstack-tox-py37/42e9995/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a616d3eadd6ae9ab8d1f5a3c8ff3d011a1b424cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:12:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dca05326d451fccb2c9fc7ae2b85deeb4cf7942","revisions":{"7dca05326d451fccb2c9fc7ae2b85deeb4cf7942":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:55:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/08/672108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/08/672108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/08/672108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/08/672108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/08/672108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/08/672108/1"}}},"commit":{"parents":[{"commit":"ebc8eb843e4c331caf9ba8b2a7692e0cc133bc97","subject":"Merge \"Enable unit tests for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/ebc8eb843e4c331caf9ba8b2a7692e0cc133bc97"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-22 16:55:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-22 16:55:07.000000000","tz":120},"subject":"Add missing space to log statement","message":"Add missing space to log statement\n\nChange-Id: I1ea64ce2217776658e8139aa041e4e217985dfbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7dca05326d451fccb2c9fc7ae2b85deeb4cf7942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7dca05326d451fccb2c9fc7ae2b85deeb4cf7942"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
