)]}'
{"id":"openstack%2Fmonasca-agent~684638","triplet_id":"openstack%2Fmonasca-agent~stable%2Ftrain~I890d1f9aff463d69ad9a2b1ddb0646f1bc579d8c","project":"openstack/monasca-agent","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I890d1f9aff463d69ad9a2b1ddb0646f1bc579d8c","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-09-25 09:05:03.000000000","updated":"2019-10-08 09:52:43.000000000","submitted":"2019-10-08 09:52:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684638-1570528363666-9cc037c6","meta_rev_id":"d717f3c77d42885baa39fd3dab50da83a72d4301","_number":684638,"virtual_id_number":684638,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"date":"2019-09-27 08:22:42.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 09:52:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-08 09:42:27.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-08 09:42:27.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 08:12:02.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-09-27 08:22:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-08 09:42:27.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-10-08 09:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"770f8cbcd58bab49d26be56bd5ee318d9511ab75","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-25 09:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f473564425fce46a88ee6936a961f35f69f6491f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 09:23:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d074914c19a4c61b06d8ad83a416273 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1259393cb41c42519eef287e272ce237 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616c198a987f419baaeaeed189758e62 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c7849b917c64ee695d4ed954643a838 : SUCCESS in 4m 28s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/51e2bd0a82194a418192680bda187b51 : FAILURE in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8a97ce278ed4945a4df14e073fae80bb9c7e5136","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-25 12:10:32.000000000","message":"Patch Set 1:\n\nHas to wait until requirements repo is branched.","accounts_in_message":[],"_revision_number":1},{"id":"46af26f2e58e0f492098ce224ebb2e5b18c49612","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-27 06:32:00.000000000","message":"Patch Set 1:\n\nWitold, better fix: see https://review.opendev.org/#/c/684639/1 and adjust the location of requirements file.\n\nWhy do you wget it at all? We can pass it in the job...","accounts_in_message":[],"_revision_number":1},{"id":"22f6856b0b07e4d548a6811c8213bc8fd8912066","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-27 08:17:24.000000000","message":"Patch Set 1:\n\n\u003e Witold, better fix: see https://review.opendev.org/#/c/684639/1 and\n \u003e adjust the location of requirements file.\n\nAgree, I think Dobroslaw will be fixing this next week. Here he started updating the scripts for monasca-log-api\nhttps://review.opendev.org/684749\n\n \u003e Why do you wget it at all? We can pass it in the job...\n\nWhat do you mean?","accounts_in_message":[],"_revision_number":1},{"id":"d6d49daac49e010bc8dc0e9989e28b6c889be18b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-27 08:22:42.000000000","message":"Patch Set 1:\n\n\u003e \u003e Why do you wget it at all? We can pass it in the job...\n \u003e \n \u003e What do you mean?\n\nFor unit tests, we pass the constraint file in via the environment variable UPPER_CONSTRAINTS_FILE, you could use the same logic.\n\nThat way in CI you always have the correct file - and even a depends-on a change that updates upper-constraints file will work nicely.","accounts_in_message":[],"_revision_number":1},{"id":"6f87501534ba99601b92bfd9caa9cd75d5495569","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-08 08:44:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7e71bac06eb63f72528771b57609d1ed72c1954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 09:40:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0107fff5eb47948df77a386f0cc884 : SUCCESS in 8m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ed9210b0ef64816894873301202555c : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72cb43d9f3564e6697f5fef14f8e8b91 : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/458f847c67e84448b17c022f1d6b7619 : SUCCESS in 4m 32s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/263d6aa56b334ea3af6fbf79042f515a : SUCCESS in 30m 15s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/d044ab9e31624736b9ba628a33abd224 : SUCCESS in 45m 41s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9c721e09be0c4e2b9665dfa40eefc152 : SUCCESS in 49m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a6b2aa55872d5eec4b150fefe57d22d8d6345d8","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-08 09:42:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWe\u0027ll update Docker image building in follow-up changes.","accounts_in_message":[],"_revision_number":1},{"id":"db529e17e658d64a8007e268d412cf123cb35b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 09:42:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bdb8715b0cdb5ec97b906b017e367b8c1bfb320","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 09:52:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"60d7adf62a3528b0fcd7be4ce06bb1a815dcafc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 09:52:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a79c0feaf194004a99a655967cad1a0 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/226524f0e9d64ee98b4beb3f5c749845 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbf7238c624f41209bb7e60b4471c14d : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d93d0cd73c541a6a65cc5c06bc37c57 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c38ffef5cfe35dc253fccdbb2654e6d86e0fb32","revisions":{"7c38ffef5cfe35dc253fccdbb2654e6d86e0fb32":{"kind":"REWORK","_number":1,"created":"2019-09-25 09:05:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/38/684638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/38/684638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/38/684638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/38/684638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/38/684638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/38/684638/1"}}},"commit":{"parents":[{"commit":"69117baed4f5233b15745624d584983dae97bd4b","subject":"Merge \"Create Docker image from Monasca repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/69117baed4f5233b15745624d584983dae97bd4b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-25 09:05:02.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-25 09:05:02.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: I890d1f9aff463d69ad9a2b1ddb0646f1bc579d8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c38ffef5cfe35dc253fccdbb2654e6d86e0fb32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c38ffef5cfe35dc253fccdbb2654e6d86e0fb32"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
