)]}'
{"id":"openstack%2Fmonasca-agent~685695","triplet_id":"openstack%2Fmonasca-agent~master~Iaab5877ac2c4b005cfc18ec7a2d6a1be448dfe6c","project":"openstack/monasca-agent","branch":"master","topic":"prometheus","hashtags":[],"change_id":"Iaab5877ac2c4b005cfc18ec7a2d6a1be448dfe6c","subject":"Add requirement for Prometheus plugin","status":"MERGED","created":"2019-09-30 12:03:45.000000000","updated":"2019-10-07 08:56:28.000000000","submitted":"2019-10-07 08:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685695-1570438588823-25c6a380","meta_rev_id":"4b211c1a46483fa904916808c4e9dcaf343b09ec","_number":685695,"virtual_id_number":685695,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 08:56:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"recommended":{"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},"all":[{"value":1,"date":"2019-10-02 14:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},{"value":2,"date":"2019-10-02 02:43:06.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-07 08:47:32.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-07 08:47:32.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 02:43:06.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-10-02 14:15:24.000000000","updated_by":{"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},"reviewer":{"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},"state":"REVIEWER"},{"updated":"2019-10-07 08:47:32.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-10-07 08:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f50ccce58f2331ecbe29d83fa7002a772a37df3e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-09-30 12:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83ff960be2cc83c683c7e5a381499f816d282c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 12:51:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/34453ffc268f4d4c877914b894b8ce8e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c256f7ecb1b24acabc2ab7eadcda8464 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cac0aabded3f4b91a133007710b7ea74 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5ce6cbfe8947c4b13e45fe2784b667 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/240411493eea42eebc4f4c35d9085678 : SUCCESS in 4m 59s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/07737e6234a947c3a9e75c884f9e11a4 : SUCCESS in 26m 03s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/55bef471ff4b48ffbbcaa050f6d7d1cd : SUCCESS in 34m 31s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b91e8930769a4f9e94e80fd5e156dbf7 : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fd8ad765112caed4c789bd5d857d5a9c5203fe3","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-02 02:43:06.000000000","message":"Patch Set 1: Code-Review+2\n\nDeployed fine on my DevStack","accounts_in_message":[],"_revision_number":1},{"id":"a15c7a2870615864878e9ed050bcd50c7d7476f0","author":{"_account_id":13348,"name":"Allan Greentree","email":"agreentree@suse.com","username":"greental"},"date":"2019-10-02 14:15:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"006fa3e86744c7075065e7974a1774696f353b57","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-10-07 08:47:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"407777f12f2ec4a11cc4b6babb91cf8ad127adc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 08:47:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee02646caa883b91fa575cfc39e392ddd5e32f47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 08:56:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5bd81a60ed783780af3800be1bdcff8ab72552bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 08:56:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51b7d53d9f3b478e95e1ba5ef708f686 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef782743b1a4f6b9c081bd807147b40 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bdfc9a054d5462c8aaef3ddec3f5404 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b64abce322d44274ac09efa31c668e86 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e6d7145467f4e73aabe50820cf91d0e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18b3098b1576bb8c4b910304c448d5e51985e8f2","revisions":{"18b3098b1576bb8c4b910304c448d5e51985e8f2":{"kind":"REWORK","_number":1,"created":"2019-09-30 12:03:45.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/95/685695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/95/685695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/95/685695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/95/685695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/95/685695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/95/685695/1"}}},"commit":{"parents":[{"commit":"c0f6f0b2f670e171dd61fd9d6cdd46aa1f9e1868","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/c0f6f0b2f670e171dd61fd9d6cdd46aa1f9e1868"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-09-30 11:57:47.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-09-30 11:57:47.000000000","tz":120},"subject":"Add requirement for Prometheus plugin","message":"Add requirement for Prometheus plugin\n\nMonasca-agent Prometheus plugin requires prometheus_client library. This\nchange adds it to extras section to simplify deploying the plugin\ntogether with its requirements.\n\nChange-Id: Iaab5877ac2c4b005cfc18ec7a2d6a1be448dfe6c\nStory: 2004268\nTask: 27807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/18b3098b1576bb8c4b910304c448d5e51985e8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/18b3098b1576bb8c4b910304c448d5e51985e8f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
