)]}'
{"id":"openstack%2Fmonasca-agent~730124","triplet_id":"openstack%2Fmonasca-agent~master~I6341f96f5c186428d2d829cabf618a6f84f40ce2","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"I6341f96f5c186428d2d829cabf618a6f84f40ce2","subject":"Fix parsing of StatsD metrics with Py3","status":"MERGED","created":"2020-05-21 20:52:51.000000000","updated":"2020-06-25 08:54:16.000000000","submitted":"2020-05-27 08:55:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730124-1590569748882-ba137570","meta_rev_id":"8771c9d7ecfc7b0be7ca3c01c93d969bf0911b06","_number":730124,"virtual_id_number":730124,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-22 20:11:59.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 08:55:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-26 16:55:55.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":1,"date":"2020-05-27 08:43:52.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 16:55:55.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-27 08:43:52.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-05-27 08:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8cb2990fa5d97c06f8d6b2517969390d6269cec","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-21 20:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6efaa8504305a7521b4f6e3a16520994d004983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d970c2150d794a5bbb5e168bab3e1f7c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9754deba394d4457a4f7c187d2ee741a : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd4ff83d24f74fbdb4f125fd972d9c88 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65fb7c3074064f629865879e9163a2c4 : SUCCESS in 7m 11s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/2a912b5db4224851aa44959647368c47 : SUCCESS in 26m 15s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/cfe602428aca4f72b0223126b6133fe1 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b9407a918c64661100f7a2b8fcb202b45337844","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-22 14:36:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for the fix and for the tests. One nit inline.","accounts_in_message":[],"_revision_number":1},{"id":"ed96ad94d4b579a7d83e96daeaedc204dd630c50","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-22 20:10:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c747b3e3e95b6774af73af4521dd0ca13383f9b8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-22 20:11:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"766c8d52473abd0dc93d502ebfa650c5fae3d21e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:56:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e24c8f78b14550ba4e347417101a22 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f4f773867da44328c29f68e0a76da49 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c39f3a5854464aa2aa0fc2a43ba87d97 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/376d774bcb4a4178b139f0c48eaef63e : SUCCESS in 5m 41s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/2154f7d26579421b9b9368ab36387e6b : FAILURE in 3m 11s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8a3478d318f4461bbbd97b3ff2e44755 : SUCCESS in 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41b0567f85946695fa4b2e2eb9acde56ef62dd1f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-26 16:48:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"15331ec17de534ae32f6406f214bb6ecef2573dd","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-26 16:55:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"095c12171087bf75802f192e4afa4116537c2bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:33:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448abc25924c4d9fa00279d40223e935 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d4dfa42a2541d5a9f922aceaccf97c : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27096d2801d84eb39ff3f48750c11e64 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77d994c864774712acda6c12b7c3d377 : SUCCESS in 6m 48s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b92e9b3497d9413abf3742de7c9d16a7 : SUCCESS in 26m 09s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/33f1285707374393909db9ff3b070a13 : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0bc37cb25e98cee4755419c6d5418ee0203dccb","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-27 08:43:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43afaf5e469081ca1434bb82dbb49edd9ae9835a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 08:44:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9efd62b5b51d4863ce7c1611aec83d3bf793805c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 08:55:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c6e3519a5650addc15cfd206a602f6c6790649f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 08:55:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed311dc4d6047db9ab887c7f1a99fcf : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3875b15ce9e345e087c6435e7fc2b520 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d84c27349c0c41dcbce5c78300195075 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0fbb2f685c4d98d829b3906f715410c4497e3fe6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 18:02:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"75602fcd4455fd2473b0938271463a8dd5fd5329","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-19 08:18:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c8072d716919487558996a252c750b41bf863f3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-19 08:36:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8771c9d7ecfc7b0be7ca3c01c93d969bf0911b06","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-25 08:54:16.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 7ff2c9cbd0324f04a96a9b115ded262ec2d7fa02","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2400fcf121a30f9fe78bef1c73e32b5eed48ea6","revisions":{"1ec41f6c9ce2a897d8e11700fe5e8ef352b803e1":{"kind":"REWORK","_number":1,"created":"2020-05-21 20:52:51.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/24/730124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/24/730124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/1"}}},"commit":{"parents":[{"commit":"579dc2c6c51233aa75d27e6481f8ee583045dde6","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/579dc2c6c51233aa75d27e6481f8ee583045dde6"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-21 20:42:20.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-21 20:52:36.000000000","tz":60},"subject":"Fix parsing of StatsD metrics with Py3","message":"Fix parsing of StatsD metrics with Py3\n\nWhen running with Py3 we compare a byte string to a unicode string\nwhen parsing StatsD metrics. This patch adds some unit tests to\nreproduce the bug and decodes the bytestring to make the existing\ncomparisons valid under Py3. When backporting to Train we can use\nOslo encodeutils. Clearly we could have more unit tests, but\nthis makes a start.\n\nChange-Id: I6341f96f5c186428d2d829cabf618a6f84f40ce2\nStory: 2007684\nTask: 39796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1ec41f6c9ce2a897d8e11700fe5e8ef352b803e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1ec41f6c9ce2a897d8e11700fe5e8ef352b803e1"}]},"branch":"refs/heads/master"},"a2400fcf121a30f9fe78bef1c73e32b5eed48ea6":{"kind":"REWORK","_number":2,"created":"2020-05-22 20:10:46.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/24/730124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/24/730124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/24/730124/2"}}},"commit":{"parents":[{"commit":"579dc2c6c51233aa75d27e6481f8ee583045dde6","subject":"Add CONTRIBUTING.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/579dc2c6c51233aa75d27e6481f8ee583045dde6"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-21 20:42:20.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-05-22 20:08:28.000000000","tz":60},"subject":"Fix parsing of StatsD metrics with Py3","message":"Fix parsing of StatsD metrics with Py3\n\nWhen running with Py3 we compare a byte string to a unicode string\nwhen parsing StatsD metrics. This patch adds some unit tests to\nreproduce the bug and decodes the bytestring to make the existing\ncomparisons valid under Py3. When backporting to Train we can use\nOslo encodeutils. Clearly we could have more unit tests, but\nthis makes a start.\n\nChange-Id: I6341f96f5c186428d2d829cabf618a6f84f40ce2\nStory: 2007684\nTask: 39796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a2400fcf121a30f9fe78bef1c73e32b5eed48ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a2400fcf121a30f9fe78bef1c73e32b5eed48ea6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
