)]}'
{"id":"openstack%2Fmonasca-agent~732128","triplet_id":"openstack%2Fmonasca-agent~master~Ifddba6aa9350f722a741a28a152ed9bc3e0b7da6","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"Ifddba6aa9350f722a741a28a152ed9bc3e0b7da6","subject":"Remove dependency on local hostname","status":"MERGED","created":"2020-06-01 09:49:53.000000000","updated":"2020-06-03 19:47:07.000000000","submitted":"2020-06-03 19:47:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732128-1591213627135-fa71b0af","meta_rev_id":"7bb008ce146cc55b162ca09d4d1862022680f439","_number":732128,"virtual_id_number":732128,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 19:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-03 19:36:23.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":2,"date":"2020-06-03 13:19:48.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-03 19:36:23.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 13:19:48.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-06-03 19:36:23.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-06-03 19:47:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d904d61cba18671740078c7289e3d2847cbc2575","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-01 09:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb1a704677e1c2eb4165b5ebcae8fc306e5ba53b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 10:37:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3673136e71b741b3b60307209080a68d : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d5409d498174af583211be750ecf26c : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2617e796add54f1bb7bb063cdf75848a : SUCCESS in 7m 15s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/665895ce549447bcab817ef408e48f21 : SUCCESS in 25m 59s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/2eed0e85d4204050919924aacd3bcb0b : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aca9b9888aeda702fe100652817c549880059e79","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-06-03 13:19:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56a6c14f5414369e0b8309caca6d095f763df9eb","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-06-03 19:36:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bcf279ceea04e6abc5e3f6a0469cb28c627226c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 19:36:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88376480c1110538396dbdb26e95c0a4b96daced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 19:47:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c34402db05143fbb4652d96bc630ad8 : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5160688b5f4e97803f6834f173658c : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d17cc139034b52bfa9ce6741c556d2 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7bb008ce146cc55b162ca09d4d1862022680f439","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 19:47:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5af9c5a4854a490d3a87ed24913b84fc7da46b70","revisions":{"5af9c5a4854a490d3a87ed24913b84fc7da46b70":{"kind":"REWORK","_number":1,"created":"2020-06-01 09:49:53.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/28/732128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/28/732128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/732128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/732128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/732128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/28/732128/1"}}},"commit":{"parents":[{"commit":"7c087801c97250f2064613135bd4aa7911c1fb91","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c087801c97250f2064613135bd4aa7911c1fb91"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-01 09:39:33.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-06-01 09:43:47.000000000","tz":60},"subject":"Remove dependency on local hostname","message":"Remove dependency on local hostname\n\nThe json_plugin tests read the localhost name in a different way\nto how the hostname is fetched in the Monasca Agent check. This can\ncause these tests to fail since the hostnames may differ. For example,\nonly one may contain .home as a suffix. In this change we avoid reading\nthe hostname from the system running the tests.\n\nStory: 2007743\nTask: 39921\nChange-Id: Ifddba6aa9350f722a741a28a152ed9bc3e0b7da6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/5af9c5a4854a490d3a87ed24913b84fc7da46b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/5af9c5a4854a490d3a87ed24913b84fc7da46b70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
