)]}'
{"id":"openstack%2Fmonasca-agent~745517","triplet_id":"openstack%2Fmonasca-agent~master~Ia3f2c8abc87cf5551d3469d616790e8e9d567bce","project":"openstack/monasca-agent","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"Ia3f2c8abc87cf5551d3469d616790e8e9d567bce","subject":"Use unittest.mock instead of mock","status":"MERGED","created":"2020-08-10 11:45:09.000000000","updated":"2020-08-10 13:44:56.000000000","submitted":"2020-08-10 13:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745517-1597067096374-453d58e8","meta_rev_id":"05e6c4e90822f3317bbc9217f4c0aa5f122acef6","_number":745517,"virtual_id_number":745517,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 13:44:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"recommended":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 12:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":2,"date":"2020-08-10 13:31:51.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"value":1,"date":"2020-08-10 13:31:51.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 12:37:13.000000000","updated_by":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2020-08-10 13:31:51.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-08-10 13:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"183d9bd0de7946cf49ab23d4fc1789a8be8743e6","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-08-10 11:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f187c260d2e2ec3e0de06369991bc68970870e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc3f0ec9c2044ba9f72210ddd34764a : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b56fdd87cd2d4c39a6be97eec4ad0715 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06dafbbb28df4e99aceb304eab5d1ac2 : SUCCESS in 4m 09s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/4558b5d1b9784b9b977395c2527a2819 : SUCCESS in 25m 40s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ee58435d4f304385bfabed72da3b605f : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f5e071bb6c3b82446c1b777c4abb81106e6c842","author":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"date":"2020-08-10 12:37:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b95a1fa14d61e5d698a91f59e89c080bcf7a964f","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-08-10 13:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6a9ea2fa31ae198704fa901d5156527a3d2110a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 13:32:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68e70cd8d0b6c7383e251c209a874ed7826f9815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 13:44:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b61cbfb43ff518537157331faba89f235f11d623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 13:44:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784bae60a94b48b69db9ed5ad56d802a : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e21e3e56a806441b8e639845576282f8 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f2ba292e9843f5b58098b2352c7cef : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3cc8c1e7d8808ca3f1b4fb94c06b9c91af5f710","revisions":{"d3cc8c1e7d8808ca3f1b4fb94c06b9c91af5f710":{"kind":"REWORK","_number":1,"created":"2020-08-10 11:45:09.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/17/745517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/17/745517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/17/745517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/17/745517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/17/745517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/17/745517/1"}}},"commit":{"parents":[{"commit":"3d0f3e24350eabc680625726a500f42a5868c337","subject":"Start README.rst with a better title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/3d0f3e24350eabc680625726a500f42a5868c337"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-08-10 11:41:08.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-08-10 11:41:08.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: Ia3f2c8abc87cf5551d3469d616790e8e9d567bce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d3cc8c1e7d8808ca3f1b4fb94c06b9c91af5f710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d3cc8c1e7d8808ca3f1b4fb94c06b9c91af5f710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
