)]}'
{"id":"openstack%2Fmonasca-agent~746371","triplet_id":"openstack%2Fmonasca-agent~master~I4a70b786ac7e2739a32616dacbc5e44d29d9c922","project":"openstack/monasca-agent","branch":"master","topic":"Story/2008017","hashtags":[],"change_id":"I4a70b786ac7e2739a32616dacbc5e44d29d9c922","subject":"Python 3: Use byte instead of string.","status":"ABANDONED","created":"2020-08-15 13:45:45.000000000","updated":"2021-06-10 12:19:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0620c04d0533c5df5d1bccda390c3803d3fdff83","_number":746371,"virtual_id_number":746371,"owner":{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-15 14:24:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-15 14:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d6f3cd0a4507a691a503077fe08bd01ca43dfd6","author":{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"},"date":"2020-08-15 13:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbea6b296ad9e94b6063be76931f0c2dccb6d067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 14:24:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a590ad7f6bd8490494a5466fc5bd878b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe785341ff444a9883ced0f47049f2c4 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61734941499428c9d3b9051f45a9242 : SUCCESS in 4m 18s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/4509fbbc7ed2466e8b3dfdb873f402eb : SUCCESS in 19m 21s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/da9790c0c9324048a51a7af851107362 : SUCCESS in 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5b2aeeea39444fb73e4c1f09bcc26b044a91007","author":{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"},"date":"2020-08-18 17:19:45.000000000","message":"Topic set to Story/2008017","accounts_in_message":[],"_revision_number":1},{"id":"0620c04d0533c5df5d1bccda390c3803d3fdff83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 12:19:43.000000000","message":"Abandoned\n\nReason described in Story: 2008017","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"543121c4d9f39ecb1d69bc4fd613ec7876f130cf","revisions":{"543121c4d9f39ecb1d69bc4fd613ec7876f130cf":{"kind":"REWORK","_number":1,"created":"2020-08-15 13:45:45.000000000","uploader":{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"},"ref":"refs/changes/71/746371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/71/746371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/71/746371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/71/746371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/71/746371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/71/746371/1"}}},"commit":{"parents":[{"commit":"caffc7fbcf4987f25dd7dd07cd0cef2df309c186","subject":"Avoid single quote in help text","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/caffc7fbcf4987f25dd7dd07cd0cef2df309c186"}]}],"author":{"name":"congtt2","email":"congtt2@vng.com.vn","date":"2020-08-15 13:41:15.000000000","tz":420},"committer":{"name":"congtt2","email":"congtt2@vng.com.vn","date":"2020-08-15 13:44:01.000000000","tz":420},"subject":"Python 3: Use byte instead of string.","message":"Python 3: Use byte instead of string.\n\nIn Python 3, output of popen.communicate is byte instead of str.\n\nStory: 2008017\nChange-Id: I4a70b786ac7e2739a32616dacbc5e44d29d9c922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/543121c4d9f39ecb1d69bc4fd613ec7876f130cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/543121c4d9f39ecb1d69bc4fd613ec7876f130cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
