)]}'
{"id":"openstack%2Fmonasca-agent~750213","triplet_id":"openstack%2Fmonasca-agent~master~I76823e7462eb31037b34a56fb61b60a002286dee","project":"openstack/monasca-agent","branch":"master","topic":"app-cred","hashtags":[],"change_id":"I76823e7462eb31037b34a56fb61b60a002286dee","subject":"Add support for application credentials","status":"ABANDONED","created":"2020-09-07 21:01:18.000000000","updated":"2025-06-05 14:37:46.000000000","total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"addf320d57d0ab4ba2143d6265364dd7e927ceed","_number":750213,"virtual_id_number":750213,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-19 14:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"all":[{"value":1,"date":"2021-03-26 16:33:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 21:33:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-08 09:16:06.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-09-08 13:30:16.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2021-03-19 14:17:05.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2021-03-23 19:36:53.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"state":"REVIEWER"},{"updated":"2021-03-30 13:28:50.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"}],"messages":[{"id":"cc3a169fde26415aec06ca3db9ebbb501639d7e3","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-09-07 21:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89b78c423ec53d88375a0c45f469633fcdf50e9e","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-09-07 21:02:16.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds expert scrutiny, test cases and documentation!","accounts_in_message":[],"_revision_number":1},{"id":"4c450c51ae7bea7670c66d32c58b64978743a1d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:33:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62611dfe5f0543ef937bf274e9bed90b : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1ebc3361d80433a82b4a36068f01352 : FAILURE in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de48ef8a71c1441abbcf3429ee47b035 : SUCCESS in 5m 28s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/91301d4d75b940d99754fc08508e5f17 : SUCCESS in 23m 37s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f520df67933f4171b65e0710d5a7d138 : SUCCESS in 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a17193aa3fbf068c9604c1868833c2e4ce9528aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-03-19 14:17:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20cc4f86af3108efc48a30aca67a46fd9e15220b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:50:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86dae0fd82104241be1a70d74c6e2369 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7f283c79f4f4dd1aecc11226e516f72 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcde7e0bc5694d20882518dc3c23d13b : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bbe28c4879d48b5b20204f90831be2d : SUCCESS in 6m 43s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/9fbbb39529ad4d42b0f90b72b7cea8fa : SUCCESS in 24m 42s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f6cf735c3c984aa1b00c7c66a9fe5987 : SUCCESS in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9c6819488d2b01813b53fd05998ce6e0774f206","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2021-03-23 19:36:53.000000000","message":"Patch Set 2:\n\nHi Sebastian, can you take a look at this patch?  It is the Monasca-Agent application credential support, with some extra unit tests added.","accounts_in_message":[],"_revision_number":2},{"id":"3fce8e7e4f1435b126fc1d436b447afa82800226","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-03-24 10:24:09.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nNice to see this, thanks! I can\u0027t see anything wrong with it, but it would be nice to make it a little more user friendly.","accounts_in_message":[],"_revision_number":2},{"id":"ca07fd474f9e9ad0dde00af5270ebc01ad61d5ab","author":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"date":"2021-03-26 16:33:02.000000000","message":"Patch Set 2: Code-Review+1\n\nHi Stig,\n\nThanks for the heads-up.\n\nIt looks good to me, and I have successfully tested it on test VMs on our infrastructure.\n\nBest regards,\nSebastian","accounts_in_message":[],"_revision_number":2},{"id":"addf320d57d0ab4ba2143d6265364dd7e927ceed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2025-06-05 14:37:46.000000000","message":"Abandoned\n\nNo further progress planned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"437546a83186a8e60dc7a5f461d26295bd9e4b4b","revisions":{"28cc1ff672be028e28a917775e6546a9dff86cff":{"kind":"REWORK","_number":1,"created":"2020-09-07 21:01:18.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/13/750213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/13/750213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/1"}}},"commit":{"parents":[{"commit":"3d0f3e24350eabc680625726a500f42a5868c337","subject":"Start README.rst with a better title","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/3d0f3e24350eabc680625726a500f42a5868c337"}]}],"author":{"name":"Stig Telfer","email":"stig.openstack@telfer.org","date":"2020-09-07 20:57:17.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig.openstack@telfer.org","date":"2020-09-07 20:59:59.000000000","tz":60},"subject":"Add support for application credentials","message":"Add support for application credentials\n\nIntroduce additional Api configuration parameters necessary\nto support authentication via application credential.\n\nChange-Id: I76823e7462eb31037b34a56fb61b60a002286dee\nStory: 2005622\nTask: 30875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/28cc1ff672be028e28a917775e6546a9dff86cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/28cc1ff672be028e28a917775e6546a9dff86cff"}]},"branch":"refs/heads/master"},"437546a83186a8e60dc7a5f461d26295bd9e4b4b":{"kind":"REWORK","_number":2,"created":"2021-03-19 14:17:05.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/13/750213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/13/750213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/13/750213/2"}}},"commit":{"parents":[{"commit":"abc8f5982ed3fc863d53b4dcb16c6855ab9e3464","subject":"Merge \"Use importlib to take place of imp module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/abc8f5982ed3fc863d53b4dcb16c6855ab9e3464"}]}],"author":{"name":"Stig Telfer","email":"stig.openstack@telfer.org","date":"2020-09-07 20:57:17.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2021-03-19 14:17:02.000000000","tz":0},"subject":"Add support for application credentials","message":"Add support for application credentials\n\nIntroduce additional Api configuration parameters necessary\nto support authentication via application credential.\n\nStory: 2005622\nTask: 30875\n\nChange-Id: I76823e7462eb31037b34a56fb61b60a002286dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/437546a83186a8e60dc7a5f461d26295bd9e4b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/437546a83186a8e60dc7a5f461d26295bd9e4b4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
