)]}'
{"id":"openstack%2Fmonasca-agent~756614","triplet_id":"openstack%2Fmonasca-agent~master~I9b7626d05b15e2f17304de0c917eea655cf5d047","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"I9b7626d05b15e2f17304de0c917eea655cf5d047","subject":"add py38 metedata","status":"MERGED","created":"2020-10-08 02:10:18.000000000","updated":"2020-10-13 16:29:02.000000000","submitted":"2020-10-13 16:29:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756614-1602606542173-577aa814","meta_rev_id":"3d1319aa8763e9f4e6eb6ba990b2eccfae45fef9","_number":756614,"virtual_id_number":756614,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-13 16:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-13 16:13:18.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-13 16:13:18.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 16:13:18.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-10-13 16:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7bbb2c09fd4ba88bb071e534f646b6039e0c45d","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2020-10-08 02:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37ea308aa79752ff954468db2c201c5a823204ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 02:44:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52ec6f62603a49358cf60f5745a27eb1 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0920731fd0824e8c9b4f788bc0495b08 : RETRY_LIMIT in 2m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/840653eb0117474b901e8d2f382edf67 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1a04e35710424d918b6cdc4e776b9b : RETRY_LIMIT in 2m 22s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/0da06ed02cb94a5888ffb1ce6a7f801c : SUCCESS in 23m 54s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/af6f826e92034d2ea8e71958b8bfe76c : FAILURE in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d9fcccb48406ec015887a4e220ed2300d008ace","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2020-10-08 07:26:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"26a107bc78dcb9624d092723c9cd90f54eafe4f1","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 14:29:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"97e8f3f85d1e63bb2780ccbbb988f7f9d1da1983","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 14:29:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9e4028b6fa8d3c6a0f22a4d53c83399704ba24bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 15:16:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81c0bbd3804249a5b362afcf447bc483 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b094080bea6d449186a90f933959759e : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47212d2d8938422786a40a2a2137647a : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d06d1121b5145fa8d3f26eed779c0e9 : SUCCESS in 5m 13s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/12c3b7dba7b04888a3ddb1d7dd358356 : SUCCESS in 25m 36s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e40ce0e1b48c4163a1e2c32537b3641e : FAILURE in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9622fd6a726a320a8ed7389fbe16a7d6e43b26b0","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 15:26:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1c65788fc55413d9b5fe409091bb089053a50ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:09:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/705f5ec5e72049f0bddaa4bb15e806c3 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bb53cc52234f93b268d2086dc7f7e0 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7a7c95cc77c42ba8e05e1cce14c574e : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf47026a167248988a326f173cc1cff4 : SUCCESS in 6m 15s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/4e1ac96656d54bcea1e829b81807a406 : SUCCESS in 25m 33s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0d563ceddc9b485e8a79e4c05e996a41 : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18cdf4a41a2ef968b6be395316e5b1f3538c938e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-13 16:13:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for the update, sorry for late review.","accounts_in_message":[],"_revision_number":2},{"id":"73f75dac81ae4bf8cb453fa7d8cae31317269acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:13:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30930ada4aa47396f9285b5124def14c65e7295b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:29:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/89d6241312804954babd9124d5f15f70 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16af2fad5e484a3192be2b223cb1bae7 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7f6130133674d5eb20fbc21534fcafc : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea7776fa55e045e0a56f9c8cd500575d : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3d1319aa8763e9f4e6eb6ba990b2eccfae45fef9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:29:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"284558499851a31844aeade923342e15deb0c90a","revisions":{"660eaedce537eb833b8ba488ef7ead4e5e25a916":{"kind":"REWORK","_number":1,"created":"2020-10-08 02:10:18.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/14/756614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/14/756614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/1"}}},"commit":{"parents":[{"commit":"e021416e600f81d5ed544b1c66675719c3a896b6","subject":"Update API version of monitored Cinder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/e021416e600f81d5ed544b1c66675719c3a896b6"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2020-10-08 02:06:56.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2020-10-08 02:06:56.000000000","tz":480},"subject":"add py38 metedata","message":"add py38 metedata\n\nChange-Id: I9b7626d05b15e2f17304de0c917eea655cf5d047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/660eaedce537eb833b8ba488ef7ead4e5e25a916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/660eaedce537eb833b8ba488ef7ead4e5e25a916"}]},"branch":"refs/heads/master"},"284558499851a31844aeade923342e15deb0c90a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-13 14:29:36.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/14/756614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/14/756614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/14/756614/2"}}},"commit":{"parents":[{"commit":"2ebd1a0755e73ff0ada5171dd90f85ae21ac9a0a","subject":"Adding python3-libvirt for Ubuntu-focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2ebd1a0755e73ff0ada5171dd90f85ae21ac9a0a"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2020-10-08 02:06:56.000000000","tz":480},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2020-10-13 14:29:36.000000000","tz":0},"subject":"add py38 metedata","message":"add py38 metedata\n\nChange-Id: I9b7626d05b15e2f17304de0c917eea655cf5d047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/284558499851a31844aeade923342e15deb0c90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/284558499851a31844aeade923342e15deb0c90a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
