)]}'
{"id":"openstack%2Fmonasca-agent~762081","triplet_id":"openstack%2Fmonasca-agent~master~I1b269b5c06a99e8f62f7c5a33b2314de06389041","project":"openstack/monasca-agent","branch":"master","topic":"assertItemsEqual","hashtags":[],"change_id":"I1b269b5c06a99e8f62f7c5a33b2314de06389041","subject":"Replace assertItemsEqual with assertCountEqual","status":"MERGED","created":"2020-11-10 06:49:20.000000000","updated":"2020-11-10 10:59:15.000000000","submitted":"2020-11-10 10:59:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762081-1605005955620-e490bc4f","meta_rev_id":"a8588a65c9acd4dd14f74532ee34cbd0cd01352a","_number":762081,"virtual_id_number":762081,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 10:59:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-10 10:47:10.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-10 10:47:10.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 10:47:10.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-11-10 10:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67f986b6446cbcf52dfcc2512721d83a6db93668","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-10 06:49:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2353cddd9a6013fcaafb387801b8f4cc75332284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 07:25:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f2f9e19257467880414b91ac906a04 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a76f8c010b144268c84e5b0cbb93edc : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd15a4f0a3c042b7b81b5cc5576b5c96 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d51edd28b0f4969a29bcd7ce997968f : SUCCESS in 6m 20s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/0a9dd3cb27484791ac9ea5c72f5ae7df : SUCCESS in 24m 45s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/14b80324e4924df4b4d0d02bdd6348b9 : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8e18e6203d64a872409d2f4197d65bd0037202e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-11-10 10:47:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nHi Likui, thanks for the change.","accounts_in_message":[],"_revision_number":1},{"id":"03315bef1d7b42ee546147a6412e0e54e6b9341c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 10:47:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9ac6fe1694a086d4e2bf087cfb5675c62622e73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 10:59:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e92714bb2aec271274ed22780d5151e1428c045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 10:59:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506a233431ff4b6f8c655524c7384009 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd585b57404d4f5e8722c88a1fd915fd : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a656d835353e4a089d6e1904b51849c3 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38c6f3af51c1ae6364531deb885af92eeee52186","revisions":{"38c6f3af51c1ae6364531deb885af92eeee52186":{"kind":"REWORK","_number":1,"created":"2020-11-10 06:49:20.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/81/762081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/81/762081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/81/762081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/81/762081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/81/762081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/81/762081/1"}}},"commit":{"parents":[{"commit":"b7179b7dc2f3486927ec1c17dc1c1051f4c2cd12","subject":"Merge \"bump py37 to py38 in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b7179b7dc2f3486927ec1c17dc1c1051f4c2cd12"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-10 06:48:15.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-10 06:48:15.000000000","tz":480},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: I1b269b5c06a99e8f62f7c5a33b2314de06389041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/38c6f3af51c1ae6364531deb885af92eeee52186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/38c6f3af51c1ae6364531deb885af92eeee52186"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
