)]}'
{"id":"openstack%2Fmonasca-agent~792379","triplet_id":"openstack%2Fmonasca-agent~master~I4d6801b2d162b2acd00d0cf07aee109ebe652718","project":"openstack/monasca-agent","branch":"master","topic":"typo-fix","hashtags":[],"change_id":"I4d6801b2d162b2acd00d0cf07aee109ebe652718","subject":"Fix typo in log message","status":"MERGED","created":"2021-05-20 13:30:00.000000000","updated":"2021-06-09 14:13:41.000000000","submitted":"2021-06-09 14:13:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792379-typo-fix","meta_rev_id":"fd7a382354f43cdafe0f563b09dbb7552890c0a9","_number":792379,"virtual_id_number":792379,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 14:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 13:59:50.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 13:59:50.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 14:16:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 11:09:40.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"CC"},{"updated":"2021-06-09 13:59:50.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"bb8ebfb96f9e98443bbfc637419a33f1ceed509d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-20 13:30:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac45a320da435b1ed17d7ae5b09710fd2252329a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-20 13:34:20.000000000","message":"Topic set to typo-fix","accounts_in_message":[],"_revision_number":1},{"id":"22ff89949f7e80f6cd046a424e9842ab39117802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 14:16:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/affdfbbd99694d8e8b8e34a5f0a6884e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64fef4ae76d34674840b91d139c4d596 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29c8f84a55ff466b8c00b669ec5fab1f : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a238687a2726462cafeeda5cb6e452ca : SUCCESS in 6m 16s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/3d1934004bdf4b9095a7c100386558e7 : FAILURE in 2m 31s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/7dbc7c7834844a78a4aa6900ae7b63b1 : FAILURE in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2db112f815641ddba417c8327203c1c2f38fd90d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-09 09:49:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f0184ece8d635cfd04930e02b1e518f40612c187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:46:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc910ae1d1db4f7e9bf20f83fcdfbb92 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/673c56008734472b8d54cd84b07a5256 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a5c98c9ce6a42f4884ca97af34da681 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06da375c35da40268225760898201067 : SUCCESS in 7m 31s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/3d82b0f27a464d748e2d986b6bcd5111 : FAILURE in 3m 15s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/12e486ea24024815a533b05792d4321f : SUCCESS in 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d1b7a591ce5ce555229c32c89f380598dce4730","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-09 11:09:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5f5cf5f0be4b5983bf7a3bf862b196273c34fa07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:46:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3167a009334d628ed04643e9a9f80e : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31121f2be25f477ca615fc410d63e4ec : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d3b6e0233a64acaa183ce68d240b2c2 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa092fb93fa744608a9172feb8f26844 : SUCCESS in 6m 43s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/33feca1ceac344ceb20f6e695a8fc897 : TIMED_OUT in 30m 51s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/901f228123544c88a28e92bc7d783ac1 : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a85594a13d23b5d4cd32d1852e652f20bf5c9ae7","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-09 12:39:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a2171fd987bac176e89a21e8836958834211ac70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:30:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737c41f84ceb41feb44509ef6daeb49f : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa81c5ca9534a1fbe35c1f9851d730f : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cacc806e9663473ea7f3a41a9f1864a0 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0be81e5b7ad94681937c0f3472b09250 : SUCCESS in 6m 17s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b2371ae38e5342e38cb0f7b62194c49e : SUCCESS in 27m 45s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0267a8b9dcf04bbba715914919c7c004 : SUCCESS in 45m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e21847448474ef7cd1adb13f06aceec4b68e3781","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-09 13:59:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for the fix","accounts_in_message":[],"_revision_number":2},{"id":"b62f1b714a2f32cca42a31d543656adec5d1a714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:00:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a015eb1ab421f2f8b57b9c9fbda7f0b8853d511b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:13:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0f86a0f21048aeb679842375355fff : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6489a99f3d4a7da96357f62492ff38 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95879a9c43e348e882f48fc9a8a12d6c : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"fd7a382354f43cdafe0f563b09dbb7552890c0a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:13:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"478a438654a011b88b529a7f91e3be0b919e76ed","revisions":{"34d95af754718f98aa74215a206d5f97e0e5eb19":{"kind":"REWORK","_number":1,"created":"2021-05-20 13:30:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/792379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/79/792379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/1"}}},"commit":{"parents":[{"commit":"cefab7f3e837188ac905a618295e3b56326e61f8","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/cefab7f3e837188ac905a618295e3b56326e61f8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-20 13:29:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-20 13:29:48.000000000","tz":120},"subject":"Fix typo in log message","message":"Fix typo in log message\n\nChange-Id: I4d6801b2d162b2acd00d0cf07aee109ebe652718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/34d95af754718f98aa74215a206d5f97e0e5eb19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/34d95af754718f98aa74215a206d5f97e0e5eb19"}]},"branch":"refs/heads/master"},"478a438654a011b88b529a7f91e3be0b919e76ed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-09 09:49:31.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/79/792379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/79/792379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/79/792379/2"}}},"commit":{"parents":[{"commit":"7d21e505e8a8f750e88978c80f7ac47d73d7f2db","subject":"Fix Docker builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7d21e505e8a8f750e88978c80f7ac47d73d7f2db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-20 13:29:48.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-09 09:49:31.000000000","tz":0},"subject":"Fix typo in log message","message":"Fix typo in log message\n\nChange-Id: I4d6801b2d162b2acd00d0cf07aee109ebe652718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/478a438654a011b88b529a7f91e3be0b919e76ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/478a438654a011b88b529a7f91e3be0b919e76ed"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
