)]}'
{"id":"openstack%2Fmonasca-agent~795764","triplet_id":"openstack%2Fmonasca-agent~master~I297c30044df9a94baa645a1a27de10bb49038440","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"I297c30044df9a94baa645a1a27de10bb49038440","subject":"Fix PY3 issue","status":"MERGED","created":"2021-06-10 11:49:03.000000000","updated":"2021-06-10 17:15:37.000000000","submitted":"2021-06-10 17:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795764","meta_rev_id":"4f0909b738d3a342034f149957943e7a23eed14d","_number":795764,"virtual_id_number":795764,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 17:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-10 14:30:03.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-10 14:42:59.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"}],"reviewers":{"REVIEWER":[{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 11:50:50.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2021-06-10 11:50:50.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"state":"REVIEWER"},{"updated":"2021-06-10 12:18:51.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":32261,"name":"Cong Tran","email":"congtt2@vng.com.vn","username":"congtt"},"state":"REVIEWER"},{"updated":"2021-06-10 12:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8db12121e3bc1ebfb737288a7dd8ba629b1065c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 11:49:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6744dfe2bf4c72e62cc9368a97edd60937c9c464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 12:06:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"76bfaf90a3c9b0ea218415fc34b5d73d30a29c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 12:48:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e1d5b32f1d407d8dcaf888ed7c196c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2484ef60d396408bb75e4f4888898d5f : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83ee979ce6b455db323d3c8587a8e35 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4221f78bca74a96983112283e975c38 : SUCCESS in 6m 42s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b1aa344fcf534629a79223a1da28b6ec : SUCCESS in 25m 01s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0fb40579282c4b55ba395bee0e612e55 : FAILURE in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67320b5c5bd457aa394ea470a0d6cc7a0c6ede7e","author":{"_account_id":16957,"name":"Matthias Bandorf","email":"Matthias.Bandorf@est.fujitsu.com","username":"mattibf"},"date":"2021-06-10 14:08:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"27dd04ca199c201fe670c86fe65e09d4722a9a94","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-06-10 14:08:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a828868a2ccc73cfc149a2570dfd48075b6718fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:09:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4dd191fc990eb636094cf4d2b7cdd32d840b474","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 14:20:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"0d91cf9ea8669778e27c6defb5a817da2134c733","tag":"autogenerated:gerrit:restore","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 14:25:31.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"d369b1b8afc11c35e4d9b3f1748d429d89376e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 14:25:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a05e3a2e9c8c29d92b5dbef1162b00b06ca4ccc2","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-06-10 14:30:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"32a46f6563f4e34542a23b812e24347d401a2f99","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-06-10 14:36:52.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"8056e2b1377e29e7274e7e4cd84327765891ade1","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2021-06-10 14:42:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8099f570f58bd054debbe8c81d8d9fac53a1803c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:12:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85a59a8536f4a23a63ecee82f9528d1 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89e21ee37b9b44d78c1f1fe4a3365bae : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ecc6710e9334ef0bf54dd39eeb94944 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023b7e325a9a453ca4afd7b3596ab5b5 : SUCCESS in 8m 09s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/9a93ba0631154c5ebd8dff73bc41027f : TIMED_OUT in 31m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8d514e7de6304c42a866415102b927b6 : FAILURE in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fd4e1bac0e123a2aee1b50bd19838a7db4daaee9","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-06-10 16:15:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"12617f74bb353a95c427249c1e98ff07f0cb1f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:03:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e20b940a384fe5b1a3dfa37541a601 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0c66eaad9b4e94b45e1d47679a64fe : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c0cf77f50e4e0788388e680e7957aa : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03da7c46be244884858c557364d48405 : SUCCESS in 6m 14s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/e1530ae069c64a56be4276b8070cc020 : SUCCESS in 26m 54s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ee463e5f19504377bb836148b0b0c440 : FAILURE in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a41cc2808e74424b6bb6e2e69a5cdf36d53277ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:03:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c394d8ac86bbc0743fc9fb759b36ea38108911e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:15:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e972129760ee4074911e393a99be2010 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1112c77d4d104117854a5584ed8bf6d3 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c46e6345d0b43bc8b25b3f445ddb759 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"4f0909b738d3a342034f149957943e7a23eed14d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:15:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"679f09a3a8f6c89b3b729b60d2174cb4ae3d891c","revisions":{"743e2d79c36943a13d9b9682cfadcd0783ee1611":{"kind":"REWORK","_number":1,"created":"2021-06-10 11:49:03.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/64/795764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/795764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/1"}}},"commit":{"parents":[{"commit":"478a438654a011b88b529a7f91e3be0b919e76ed","subject":"Fix typo in log message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/478a438654a011b88b529a7f91e3be0b919e76ed"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 11:43:56.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 11:43:56.000000000","tz":120},"subject":"Fix PY3 issue","message":"Fix PY3 issue\n\n- In PY3 the output of subprocess.Popen is a type bytes, but\nstrip() method requires a type str.\n\nChange-Id: I297c30044df9a94baa645a1a27de10bb49038440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/743e2d79c36943a13d9b9682cfadcd0783ee1611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/743e2d79c36943a13d9b9682cfadcd0783ee1611"}]},"branch":"refs/heads/master"},"1a5022d1bd029831af2e9c306c3d373e59c4ff26":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-10 12:06:59.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/64/795764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/795764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/2"}}},"commit":{"parents":[{"commit":"478a438654a011b88b529a7f91e3be0b919e76ed","subject":"Fix typo in log message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/478a438654a011b88b529a7f91e3be0b919e76ed"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 11:43:56.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 12:06:00.000000000","tz":120},"subject":"Fix PY3 issue","message":"Fix PY3 issue\n\n- In PY3 the output of subprocess.Popen is a type bytes, but\nstrip() method requires a type str.\n\nStory: 2008017\nTask: 40669\n\nChange-Id: I297c30044df9a94baa645a1a27de10bb49038440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1a5022d1bd029831af2e9c306c3d373e59c4ff26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/1a5022d1bd029831af2e9c306c3d373e59c4ff26"}]},"branch":"refs/heads/master"},"679f09a3a8f6c89b3b729b60d2174cb4ae3d891c":{"kind":"REWORK","_number":3,"created":"2021-06-10 14:25:52.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/64/795764/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/795764/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/795764/3"}}},"commit":{"parents":[{"commit":"478a438654a011b88b529a7f91e3be0b919e76ed","subject":"Fix typo in log message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/478a438654a011b88b529a7f91e3be0b919e76ed"}]}],"author":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 11:43:56.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2021-06-10 14:25:06.000000000","tz":120},"subject":"Fix PY3 issue","message":"Fix PY3 issue\n\n- In PY3 the output of subprocess.Popen is a type bytes, but\nstrip() method requires a type str.\n\nStory: 2008017\nTask: 40669\n\nChange-Id: I297c30044df9a94baa645a1a27de10bb49038440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/679f09a3a8f6c89b3b729b60d2174cb4ae3d891c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/679f09a3a8f6c89b3b729b60d2174cb4ae3d891c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
