)]}'
{"id":"openstack%2Fmonasca-agent~810115","triplet_id":"openstack%2Fmonasca-agent~master~I232506e19f593b43cc011c6499242847c9bbc108","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"I232506e19f593b43cc011c6499242847c9bbc108","subject":"Add a condition for systemd service detection in monasca_setup","status":"MERGED","created":"2021-09-20 19:06:40.000000000","updated":"2021-09-22 14:52:51.000000000","submitted":"2021-09-22 14:52:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810115","meta_rev_id":"295f413e50836998572140ea9bc340bd6e6a24e3","_number":810115,"virtual_id_number":810115,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 14:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-22 13:59:33.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-22 13:59:33.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 19:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-21 15:41:46.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"a5313639b4f8b16689053277322f435487aa4e1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-09-20 19:06:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26e518afc1adf465ead098ae77faeabbbfeb14b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 19:48:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d536ec23844248a4d694cf74c8a536 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6390ed79324586ac4aabdb065e5a7e : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d781ed7a4de4ad5be75247c4ab9d09c : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0196ec94dbaa4af2912f48330f2bb4b1 : SUCCESS in 7m 16s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/e8ab5c39ef1b4b48b7e68593038dadae : TIMED_OUT in 30m 59s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/e827236c6f5b4db1b57f1b79a0170c47 : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c38da2745a70a5050ce42657007624344e6580d7","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-09-22 08:21:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"630f09ec48b2e7b2a359424a20aad436cf5f7da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 09:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07198922005d4f20a803c5f652103d1d : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/394154e422d34e1c816a67a290bd3e99 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f80d71bdbad04377b4f847b67aed1373 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/524a6220a8a64dd38a0ca4d8607adc58 : SUCCESS in 7m 44s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/5b17b042e8f8427a87b152969f047754 : SUCCESS in 22m 14s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/185a6156b4a847afa5b9e640259c4d3e : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c46948ede088f3559d461831e8079ce89ac8193","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-09-22 13:59:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec740faac3577bd8055beb25560af19984beb0d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:43:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00cfdbe0e25688617068f60c4176ea546abcdfad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:52:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d8ed13aec24c35af6b649ae45d42dc : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b19788c81f470e849de1a0fd33dd59 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86a696141d0544699483a7dc6f9448b7 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"295f413e50836998572140ea9bc340bd6e6a24e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 14:52:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8be13bdeb84bd11019f8e0010985181115cc73a","revisions":{"a8be13bdeb84bd11019f8e0010985181115cc73a":{"kind":"REWORK","_number":1,"created":"2021-09-20 19:06:40.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/15/810115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/15/810115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/15/810115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/15/810115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/15/810115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/15/810115/1"}}},"commit":{"parents":[{"commit":"d30e12fde8a6d0923b44a015fd83a5c8060404e8","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d30e12fde8a6d0923b44a015fd83a5c8060404e8"}]}],"author":{"name":"root","email":"okozachenko1203@gmail.com","date":"2021-09-20 19:06:27.000000000","tz":120},"committer":{"name":"root","email":"okozachenko1203@gmail.com","date":"2021-09-20 19:06:27.000000000","tz":120},"subject":"Add a condition for systemd service detection in monasca_setup","message":"Add a condition for systemd service detection in monasca_setup\n\nSkip agent service detection if only installing plugins.\nThis will make it possible to use monasca-setup on the environments\n where monasca-agent is not running as systemd service.\nTypical usecase is running agent on kubernetes.\n\nChange-Id: I232506e19f593b43cc011c6499242847c9bbc108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a8be13bdeb84bd11019f8e0010985181115cc73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a8be13bdeb84bd11019f8e0010985181115cc73a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
