)]}'
{"id":"openstack%2Fmonasca-agent~811053","triplet_id":"openstack%2Fmonasca-agent~master~I906cf1ad1530de750c71cd622317d849347786ca","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"I906cf1ad1530de750c71cd622317d849347786ca","subject":"Clean up extra spaces","status":"MERGED","created":"2021-09-27 02:52:58.000000000","updated":"2021-09-30 16:21:54.000000000","submitted":"2021-09-30 16:21:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811053","meta_rev_id":"b572f494be44bbc2f52a9fc4efa997420d7d6062","_number":811053,"virtual_id_number":811053,"owner":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 16:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 16:09:36.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 16:09:36.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 03:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 16:09:36.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"fce2f6869163d8497e01bfa09d498cadcac2d2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2021-09-27 02:52:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37b798c2461a05aa50f42fe0bfaed5d307e6e849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 03:41:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b23bb193f64f3e8fed5cb84f3a89a1 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c56ab0ef4c644f6296d11d4c62a04e49 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5bc2b6d7c84ac59e78aafa89f4f07e : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f77a3ce147034d279f5fecbaa6ef4828 : SUCCESS in 6m 05s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/56aebe4e56f14a9081c10ace39907ff3 : SUCCESS in 27m 22s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0eac4c2afd1f4d478071e0a8809589fa : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a62d8c482858db2ec95d04535c5d3b4854742703","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2021-09-30 16:09:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a09f116ce791cba4e4450b3b7624f1974508d38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:13:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57646fe36d00dab87810d4bfae450886ba4b290d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:21:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba210977adae4edbabb2abd14d95892d : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d50a4c2f4748de8b992209674aca3e : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24ba10ba67ac4654aed7942661cf30b8 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b572f494be44bbc2f52a9fc4efa997420d7d6062","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:21:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c8a85a1d89dc430f831d31a36d2e282d1c2a387","revisions":{"8c8a85a1d89dc430f831d31a36d2e282d1c2a387":{"kind":"REWORK","_number":1,"created":"2021-09-27 02:52:58.000000000","uploader":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"ref":"refs/changes/53/811053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/53/811053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/53/811053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/53/811053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/53/811053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/53/811053/1"}}},"commit":{"parents":[{"commit":"a8be13bdeb84bd11019f8e0010985181115cc73a","subject":"Add a condition for systemd service detection in monasca_setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a8be13bdeb84bd11019f8e0010985181115cc73a"}]}],"author":{"name":"jayonlau","email":"jayonlau@gmail.com","date":"2021-09-27 02:51:47.000000000","tz":-240},"committer":{"name":"jayonlau","email":"jayonlau@gmail.com","date":"2021-09-27 02:52:47.000000000","tz":-240},"subject":"Clean up extra spaces","message":"Clean up extra spaces\n\nAlthough these errors are not important, they affect the code specification.\n\nChange-Id: I906cf1ad1530de750c71cd622317d849347786ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8c8a85a1d89dc430f831d31a36d2e282d1c2a387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8c8a85a1d89dc430f831d31a36d2e282d1c2a387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
