)]}'
{"id":"openstack%2Fmonasca-agent~824231","triplet_id":"openstack%2Fmonasca-agent~master~Id47ba9397e7fef1ac8622abb2a1691a260f4bc9c","project":"openstack/monasca-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"12907":{"account":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"last_update":"2022-01-27 10:08:41.000000000","reason":"Change was submitted"},"26141":{"account":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"last_update":"2022-01-27 10:08:41.000000000","reason":"Change was submitted"},"28062":{"account":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"last_update":"2022-01-27 09:48:47.000000000","reason":"Martin Chacon Piza replied on the change"}},"hashtags":[],"change_id":"Id47ba9397e7fef1ac8622abb2a1691a260f4bc9c","subject":"Remove JMSAppender.class to avoid CVE-2021-4104,","status":"MERGED","created":"2022-01-11 20:23:26.000000000","updated":"2022-01-27 10:08:41.000000000","submitted":"2022-01-27 10:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824231","meta_rev_id":"3360af32a6f896023188f8f18ccd1d02384cf2ab","_number":824231,"virtual_id_number":824231,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 10:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 09:48:47.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 09:48:47.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 21:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 15:07:05.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2022-01-25 21:29:01.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"}],"messages":[{"id":"8372a570501fdf37bc896bc750f9cf51727e25bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2022-01-11 20:23:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ea6971b10053984cfcfa5576b09b008be106a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 21:08:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2ba028a0c54dccba3bc3afa8575b82 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78a0e083c009479cacb75589bf35d0d1 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04598b9daab445fbed607f72cd5ec23 : SUCCESS in 5m 24s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/033ba5c5899e492ab02c36311c587721 : FAILURE in 2m 21s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/014e07cca50c4563b568673661b72fdb : FAILURE in 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da4190297f38571d41c12adf7b6b35799b80f62e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2022-01-21 05:45:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2271303f87da81001a0c1996d39fd312ec4137e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 06:42:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f5edd9fa4c44f8fb5d66a8ccf3525f8 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276c0aaa9645498fa55c1c0b42830db1 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/632d331c2aec48009d76996f56f26457 : SUCCESS in 5m 36s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/5ba39c6987fe430bb5618f20ef38ece3 : FAILURE in 9m 53s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/81ae2952bd8a4929a0a657de3d4bc706 : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4dfa4d6a30c169c32007505b40a2315fbb86fa68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-01-27 09:05:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1c02a98356ddb98576411a74c48695168edf252a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 09:44:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1633b26aaa94777a5e4d124cf6b329a : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/975f3f6a3fc44e69ac4e4fd5fc18b704 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3037ee488fa84fcb9128259701b97f21 : SUCCESS in 6m 21s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/18bb46e3f85e4b6f8d3fc2b743db4158 : SUCCESS in 13m 15s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/261dbb25bdcc431ebb92c5265e0c6bca : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e63ccbbe96c691d93f30fb8b2ef007748069971","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-01-27 09:48:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3646aaa66704e8b67ae36adc3df0bb0327ca4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 09:49:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83142596c1db8f019017ddfd95a2279e31c68536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:08:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafe194c03cb4e07b22f6d2f6234794b : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4197eadb61f4b9888e936ae7c18945c : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74bf42d6a510415f836d8baf641d3854 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"3360af32a6f896023188f8f18ccd1d02384cf2ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:08:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dbb766218e3d1e99a965d26e03ba3872ea6aa339","revisions":{"c44e6489149f61f6b5bbaf767d29421fd4170e5e":{"kind":"REWORK","_number":1,"created":"2022-01-11 20:23:26.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/31/824231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/31/824231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/1"}}},"commit":{"parents":[{"commit":"8c8a85a1d89dc430f831d31a36d2e282d1c2a387","subject":"Clean up extra spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8c8a85a1d89dc430f831d31a36d2e282d1c2a387"}]}],"author":{"name":"Jan Zerebecki","email":"jan.suse@zerebecki.de","date":"2022-01-11 20:15:01.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jan.suse@zerebecki.de","date":"2022-01-11 20:23:01.000000000","tz":60},"subject":"Remove JMSAppender.class to avoid CVE-2021-4104","message":"Remove JMSAppender.class to avoid CVE-2021-4104\n\nThough it does not contain a vulnerable configuration of log4j, to avoid\nneeding to prove that and false positives of security scanners, this\ncommit is the result of running the following commands:\n\nzip -q -d monasca_agent/collector/checks/libs/jmxfetch-0.3.0-jar-with-dependencies.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class\nunzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\nzip -q -d WORLDS-INF/lib/log4j.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class\nzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\n\nChange-Id: Id47ba9397e7fef1ac8622abb2a1691a260f4bc9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/c44e6489149f61f6b5bbaf767d29421fd4170e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/c44e6489149f61f6b5bbaf767d29421fd4170e5e"}]},"branch":"refs/heads/master"},"004aedbe650917a9585a91b9d3685b44bebb07fe":{"kind":"REWORK","_number":2,"created":"2022-01-21 05:45:16.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/31/824231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/31/824231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/2"}}},"commit":{"parents":[{"commit":"8c8a85a1d89dc430f831d31a36d2e282d1c2a387","subject":"Clean up extra spaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/8c8a85a1d89dc430f831d31a36d2e282d1c2a387"}]}],"author":{"name":"Jan Zerebecki","email":"jan.suse@zerebecki.de","date":"2022-01-11 20:15:01.000000000","tz":60},"committer":{"name":"Jan Zerebecki","email":"jan.suse@zerebecki.de","date":"2022-01-21 05:42:54.000000000","tz":60},"subject":"Remove JMSAppender.class to avoid CVE-2021-4104,","message":"Remove JMSAppender.class to avoid CVE-2021-4104,\n\nCVE-2022-23302, CVE-2022-23305, and CVE-2022-23307.\n\nThough it does not contain a vulnerable configuration of log4j, to avoid\nneeding to prove that and false positives of security scanners, this\ncommit is the result of running the following commands:\n\nzip -q -d monasca_agent/collector/checks/libs/jmxfetch-0.3.0-jar-with-dependencies.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class org/apache/log4j/jdbc/JDBCAppender.class org/apache/log4j/net/JMSSink.class org/apache/log4j/chainsaw\"*\"\nunzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\nzip -q -d WORLDS-INF/lib/log4j.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class org/apache/log4j/jdbc/JDBCAppender.class org/apache/log4j/net/JMSSink.class org/apache/log4j/chainsaw\"*\"\nzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\n\nChange-Id: Id47ba9397e7fef1ac8622abb2a1691a260f4bc9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/004aedbe650917a9585a91b9d3685b44bebb07fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/004aedbe650917a9585a91b9d3685b44bebb07fe"}]},"branch":"refs/heads/master"},"dbb766218e3d1e99a965d26e03ba3872ea6aa339":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-27 09:05:15.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/31/824231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/31/824231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/31/824231/3"}}},"commit":{"parents":[{"commit":"052ab23048e4eee4901f7e1b18633f69d5c7f89e","subject":"Remove py2-psutil from apk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/052ab23048e4eee4901f7e1b18633f69d5c7f89e"}]}],"author":{"name":"Jan Zerebecki","email":"jan.suse@zerebecki.de","date":"2022-01-11 20:15:01.000000000","tz":60},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-01-27 09:05:15.000000000","tz":0},"subject":"Remove JMSAppender.class to avoid CVE-2021-4104,","message":"Remove JMSAppender.class to avoid CVE-2021-4104,\n\nCVE-2022-23302, CVE-2022-23305, and CVE-2022-23307.\n\nThough it does not contain a vulnerable configuration of log4j, to avoid\nneeding to prove that and false positives of security scanners, this\ncommit is the result of running the following commands:\n\nzip -q -d monasca_agent/collector/checks/libs/jmxfetch-0.3.0-jar-with-dependencies.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class org/apache/log4j/jdbc/JDBCAppender.class org/apache/log4j/net/JMSSink.class org/apache/log4j/chainsaw\"*\"\nunzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\nzip -q -d WORLDS-INF/lib/log4j.jar org/apache/logging/log4j/core/lookup/JndiLookup.class org/apache/log4j/net/JMSAppender.class org/apache/log4j/jdbc/JDBCAppender.class org/apache/log4j/net/JMSSink.class org/apache/log4j/chainsaw\"*\"\nzip monasca_agent/collector/checks/libs/jmxterm-1.0-DATADOG-uber.jar WORLDS-INF/lib/log4j.jar\n\nChange-Id: Id47ba9397e7fef1ac8622abb2a1691a260f4bc9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dbb766218e3d1e99a965d26e03ba3872ea6aa339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dbb766218e3d1e99a965d26e03ba3872ea6aa339"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
