)]}'
{"id":"openstack%2Fmonasca-agent~831516","triplet_id":"openstack%2Fmonasca-agent~master~Idb98e3b64528bb5ad12e7292bfb18c159585f87b","project":"openstack/monasca-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"33881":{"account":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"last_update":"2022-04-04 10:43:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb98e3b64528bb5ad12e7292bfb18c159585f87b","subject":"Update python testing classifier","status":"MERGED","created":"2022-03-02 12:06:04.000000000","updated":"2022-04-04 10:43:36.000000000","submitted":"2022-04-04 10:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831516","meta_rev_id":"ea6882eca10335ba655ec76cf05d34f5a6e7c816","_number":831516,"virtual_id_number":831516,"owner":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 10:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-04 10:30:58.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-04 10:30:58.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 12:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 10:30:58.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"2d58708027e393be16fa4ca3255bfb4e824e67e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-03-02 12:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db2d9704da2711ae4b3fa560b0b0d1beca38e56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"date":"2022-03-02 12:10:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b3310d4e16edd1747954f62d202132b9fd5966fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:49:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f1abcbe076b46d782716e27335bc6a7 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a1e9e3cbe54fe7be708d938df08c05 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b48c6989d12e4f8ca2debbb357a5d876 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5016e2a8813044b482b52bac8ea3be4c : SUCCESS in 7m 27s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/50d0685fb30840208dba61d93fc8c532 : SUCCESS in 16m 31s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/7442dc4268d84151800ca5ca6aa0319e : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f1dd343bc663edc72a01db948580f7aed2218d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-04-04 09:20:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3c2a8af437e9467e8a6223fec062f9021a8fcda2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:06:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e38347bac013494a8dc88a74dda4382c : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30bb68f16fa4299a4b2ae8adce6d1cb : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/919c0ee0573c4822b866784f11355fa0 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48bc58fdee5048dc9610831ef0d2f4dd : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37c52778fe834dcba051cf2b4cf2bd8e : RETRY_LIMIT in 3m 26s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/15d43c85bbb54f7c807f07c03ec0f01e : SUCCESS in 15m 14s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/1ae6f772622d4783a13301af650a27ba : SUCCESS in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62673f8b9b351ec3c653a90d7e270a0dd1bf3d68","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-04-04 10:30:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3216217297a1320d6229aba00d244838801cbdab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:31:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0fc22b9763ff8d707b456e5fe76f99430e141c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:43:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7dd6dd43bbf44bfd97bc93884551faeb : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d5e8f2fd5844ceb4558c9142560b46 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e646ced756844218999cf249f26105a : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44b712ba81c84f47af03359a3e41fcc8 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"ea6882eca10335ba655ec76cf05d34f5a6e7c816","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:43:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","revisions":{"754c4b03672b371174109ca86cb759ed288870d7":{"kind":"REWORK","_number":1,"created":"2022-03-02 12:06:04.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/16/831516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/16/831516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/1"}}},"commit":{"parents":[{"commit":"4273b124fc7660fb229e81e45be42f5b0e49410c","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4273b124fc7660fb229e81e45be42f5b0e49410c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-03-02 11:59:02.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-03-02 11:59:02.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nChange-Id: Idb98e3b64528bb5ad12e7292bfb18c159585f87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/754c4b03672b371174109ca86cb759ed288870d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/754c4b03672b371174109ca86cb759ed288870d7"}]},"branch":"refs/heads/master"},"49185fb4c18561b2ce9f8bc21d85b35877cf5c5e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-02 12:10:10.000000000","uploader":{"_account_id":33881,"name":"LiZekun","email":"2954674728@qq.com","username":"joeylee666666"},"ref":"refs/changes/16/831516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/16/831516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/2"}}},"commit":{"parents":[{"commit":"4273b124fc7660fb229e81e45be42f5b0e49410c","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4273b124fc7660fb229e81e45be42f5b0e49410c"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-03-02 11:59:02.000000000","tz":480},"committer":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-03-02 12:10:10.000000000","tz":0},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Idb98e3b64528bb5ad12e7292bfb18c159585f87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/49185fb4c18561b2ce9f8bc21d85b35877cf5c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/49185fb4c18561b2ce9f8bc21d85b35877cf5c5e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"29be6e96a6c590221e7d2fd1e30dffc7aec519a9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-04 09:20:29.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/16/831516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/16/831516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/16/831516/3"}}},"commit":{"parents":[{"commit":"9ac8a8e930de389f3be54fc82b44df96c0c06fd0","subject":"Merge \"Use TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/9ac8a8e930de389f3be54fc82b44df96c0c06fd0"}]}],"author":{"name":"LiZekun","email":"2954674728@qq.com","date":"2022-03-02 11:59:02.000000000","tz":480},"committer":{"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","date":"2022-04-04 09:20:29.000000000","tz":0},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Idb98e3b64528bb5ad12e7292bfb18c159585f87b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
