)]}'
{"id":"openstack%2Fmonasca-agent~831522","triplet_id":"openstack%2Fmonasca-agent~master~Ica3df645df3b664e58774c55b461ea66be662275","project":"openstack/monasca-agent","branch":"master","hashtags":[],"change_id":"Ica3df645df3b664e58774c55b461ea66be662275","subject":"[Test] how container image is built","status":"ABANDONED","created":"2022-03-02 13:05:39.000000000","updated":"2022-03-03 15:49:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e376820c3a0d840d6c5b69a407c0a75e6f3301e","_number":831522,"virtual_id_number":831522,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-02 13:43:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 13:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa52f30154bf1a2d4526cf63347b2c7cb7675c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-03-02 13:05:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb30f30373488888f48ef6c508811f585a49a4cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:43:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9cc1b417634e4fa0736e68845e7beb : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8da4adfe8e4c27a990551bedfca567 : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/658659e54ea54e909925c769d90b47f1 : SUCCESS in 6m 44s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/bebb795ab1014cab931f436d5260310b : SUCCESS in 13m 09s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/122c1cb98a0e45b9bcd2c57c009b6155 : SUCCESS in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e376820c3a0d840d6c5b69a407c0a75e6f3301e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-03-03 15:49:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a95d42106f114c0f86d32806d9ae19651d6afd57","revisions":{"a95d42106f114c0f86d32806d9ae19651d6afd57":{"kind":"REWORK","_number":1,"created":"2022-03-02 13:05:39.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/22/831522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/22/831522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/22/831522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/22/831522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/22/831522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/22/831522/1"}}},"commit":{"parents":[{"commit":"dbb766218e3d1e99a965d26e03ba3872ea6aa339","subject":"Remove JMSAppender.class to avoid CVE-2021-4104,","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dbb766218e3d1e99a965d26e03ba3872ea6aa339"}]}],"author":{"name":"root","email":"okozachenko1203@gmail.com","date":"2022-03-02 13:05:05.000000000","tz":660},"committer":{"name":"root","email":"okozachenko1203@gmail.com","date":"2022-03-02 13:05:05.000000000","tz":660},"subject":"[Test] how container image is built","message":"[Test] how container image is built\n\nChange-Id: Ica3df645df3b664e58774c55b461ea66be662275\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a95d42106f114c0f86d32806d9ae19651d6afd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a95d42106f114c0f86d32806d9ae19651d6afd57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
