)]}'
{"id":"openstack%2Fmonasca-agent~833128","triplet_id":"openstack%2Fmonasca-agent~master~Idbde83566af37d8c35d71e280a385a16bc9f8843","project":"openstack/monasca-agent","branch":"master","topic":"add-zed-python-jobtemplates","hashtags":[],"change_id":"Idbde83566af37d8c35d71e280a385a16bc9f8843","subject":"Add Python3 zed unit tests","status":"MERGED","created":"2022-03-10 17:06:23.000000000","updated":"2022-03-11 14:19:45.000000000","submitted":"2022-03-11 14:19:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833128-add-zed-python-jobtemplates","meta_rev_id":"0a386e40cef9be27ee109fa98059f015b08e6df9","_number":833128,"virtual_id_number":833128,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 14:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-11 14:08:13.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-11 14:08:13.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 18:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 14:08:13.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"59b158081b5632be83cd3a5375d81c93e2bbdf4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-10 17:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05400886070b1eceed24d5b91b74134d066f188d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:08:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa2a7f08a4d443f999c333993fec439 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804b4cab972d45c2909a95f4d960d4a2 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab7da10063354be9842983f312dc3099 : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ed5c93f4dd84e259b14e4465c47f83f : RETRY_LIMIT in 5m 54s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/819d0dd5f50640dfaf4302d4436fd7ce : SUCCESS in 19m 18s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/40713a930f1447c7bdf5208c104021ff : FAILURE in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"966ccfc7deac01de4b773d19c65b2de22424ab4e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2022-03-11 14:08:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef8a827c4abc13f5e5b886b68f07d5284bb690fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee5cfce5e4dc22107a71240db1736bc1a7a4f9a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:19:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67970b3f8b3e40849a01c8f7323502d2 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93ccf8a25d44965a1dfe1b13f17a017 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebc42ec9fb4f488c92685dc9ccabf46c : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0a386e40cef9be27ee109fa98059f015b08e6df9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:19:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f26337fc04a4e12f265b29be80ac619c251d2851","revisions":{"f26337fc04a4e12f265b29be80ac619c251d2851":{"kind":"REWORK","_number":1,"created":"2022-03-10 17:06:23.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/28/833128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/28/833128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/833128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/833128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/28/833128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/28/833128/1"}}},"commit":{"parents":[{"commit":"dbb766218e3d1e99a965d26e03ba3872ea6aa339","subject":"Remove JMSAppender.class to avoid CVE-2021-4104,","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/dbb766218e3d1e99a965d26e03ba3872ea6aa339"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-10 17:06:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-10 17:06:22.000000000","tz":0},"subject":"Add Python3 zed unit tests","message":"Add Python3 zed unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for zed.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Idbde83566af37d8c35d71e280a385a16bc9f8843\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/f26337fc04a4e12f265b29be80ac619c251d2851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/f26337fc04a4e12f265b29be80ac619c251d2851"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
