)]}'
{"id":"openstack%2Fmonasca-agent~858197","triplet_id":"openstack%2Fmonasca-agent~master~Ie08300ae4607eff0d31982fabd73ec6cd575548e","project":"openstack/monasca-agent","branch":"master","topic":"add-antelope-python-jobtemplates","hashtags":[],"change_id":"Ie08300ae4607eff0d31982fabd73ec6cd575548e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","status":"ABANDONED","created":"2022-09-16 15:32:24.000000000","updated":"2026-02-03 22:21:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5019ad5d4200b6925a5d3ed77f9a2d4f4b6c2a5","_number":858197,"virtual_id_number":858197,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-16 16:21:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 16:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:21:24.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"5f89eb9c1ca6523a9222e77bca1781f5c2ed78d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-16 15:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d113938c6ac17cdd716b1bc87701c553bbd82fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:21:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b9d80f866a34b62aa6e62649effde71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0c880520d0450a92d30030148942b4 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d698cf302374315b74772f6cd8f56e8 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00e9236dfbee48d8967b1e15b48f442f : RETRY_LIMIT in 2m 38s\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/9b4fce769e7c4a25955964e761ad19cc : FAILURE in 5m 33s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d11b591f95e54447a25a85d35eb7b2fb : SUCCESS in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5019ad5d4200b6925a5d3ed77f9a2d4f4b6c2a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:21:24.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c9c70a417b09f593d1bb9f608a867c605e386d5","revisions":{"2c9c70a417b09f593d1bb9f608a867c605e386d5":{"kind":"REWORK","_number":1,"created":"2022-09-16 15:32:24.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/858197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/97/858197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/97/858197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/97/858197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/97/858197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/97/858197/1"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:32:21.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-16 15:32:21.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie08300ae4607eff0d31982fabd73ec6cd575548e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2c9c70a417b09f593d1bb9f608a867c605e386d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/2c9c70a417b09f593d1bb9f608a867c605e386d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
