)]}'
{"id":"openstack%2Fmonasca-agent~868865","triplet_id":"openstack%2Fmonasca-agent~master~Idf2e07014153cf20df5ac1151573dbe4a615b4b0","project":"openstack/monasca-agent","branch":"master","topic":"fix_prometheus_pod_autodiscovery","attention_set":{},"removed_from_attention_set":{"35572":{"account":{"_account_id":35572,"name":"Pawel Kubica","email":"pawel.kubica@comarch.com","username":"pawcykca"},"last_update":"2026-02-03 22:21:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf2e07014153cf20df5ac1151573dbe4a615b4b0","subject":"Fix autodiscovery using pods annotations in prometheus plugin.","status":"ABANDONED","created":"2022-12-29 11:56:40.000000000","updated":"2026-02-03 22:21:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0241527f11b39ec5d0a3a209d8990fc1504e6388","_number":868865,"virtual_id_number":868865,"owner":{"_account_id":35572,"name":"Pawel Kubica","email":"pawel.kubica@comarch.com","username":"pawcykca"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-29 12:27:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-29 12:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-03 22:21:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"5bb7b6ae937aa8edcf8915f9c9bb2c9123527558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35572,"name":"Pawel Kubica","email":"pawel.kubica@comarch.com","username":"pawcykca"},"date":"2022-12-29 11:56:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff3a82ff4f8b61216f044f95c8a8c72020f024fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 12:27:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70e581751040482bacf9ff746ce52dc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdc062b030240dfa74dad2e1feb3f67 : RETRY_LIMIT in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eac6747444dd46e88a9c290d6b1d4612 : FAILURE in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad7b7742e0e74f88b49c4ce94897ff1c : FAILURE in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/163269e0a8fc4a8687a63d4a15a45364 : RETRY_LIMIT in 1m 38s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/a5f91decc307491785f3896ef4b407f2 : FAILURE in 4m 06s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/18d4f320ea1749fabd46a1db419e7ee5 : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0241527f11b39ec5d0a3a209d8990fc1504e6388","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:21:03.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9b6edd093ccd97de78c49f38ca469164cc5c1c4","revisions":{"a9b6edd093ccd97de78c49f38ca469164cc5c1c4":{"kind":"REWORK","_number":1,"created":"2022-12-29 11:56:40.000000000","uploader":{"_account_id":35572,"name":"Pawel Kubica","email":"pawel.kubica@comarch.com","username":"pawcykca"},"ref":"refs/changes/65/868865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/65/868865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/65/868865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/65/868865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/65/868865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/65/868865/1"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Paweł Kubica","email":"pawel.kubica@comarch.com","date":"2022-12-29 11:54:11.000000000","tz":60},"committer":{"name":"Paweł Kubica","email":"pawel.kubica@comarch.com","date":"2022-12-29 11:54:11.000000000","tz":60},"subject":"Fix autodiscovery using pods annotations in prometheus plugin.","message":"Fix autodiscovery using pods annotations in prometheus plugin.\n\nUsing Kubernetes API endpoint instaed of kubelet readonly endpoint to get all pods and their annotations.\nUpdate API endpoint to get replicaset details.\n\nTask: 47092\nStory: 2010505\nSigned-off-by: Paweł Kubica \u003cpawel.kubica@comarch.com\u003e\nChange-Id: Idf2e07014153cf20df5ac1151573dbe4a615b4b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a9b6edd093ccd97de78c49f38ca469164cc5c1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a9b6edd093ccd97de78c49f38ca469164cc5c1c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
