)]}'
{"id":"openstack%2Fmonasca-agent~915964","triplet_id":"openstack%2Fmonasca-agent~master~Iddf3469f07486f2c58b46ea993547eecd6df3382","project":"openstack/monasca-agent","branch":"master","attention_set":{},"removed_from_attention_set":{"35988":{"account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"last_update":"2024-04-16 12:56:34.000000000","reason":"\u003cGERRIT_ACCOUNT_35988\u003e replied on the change","reason_account":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}}},"hashtags":[],"change_id":"Iddf3469f07486f2c58b46ea993547eecd6df3382","subject":"fix: Making the gates happy","status":"MERGED","created":"2024-04-16 06:23:48.000000000","updated":"2024-04-16 13:12:47.000000000","submitted":"2024-04-16 13:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915964","meta_rev_id":"1eebc80237bec5c50e351434b75a1dd6da30c5f0","_number":915964,"virtual_id_number":915964,"owner":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 13:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":2,"date":"2024-04-16 12:53:20.000000000","_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2024-04-16 12:56:34.000000000","_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 08:14:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 12:53:20.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"}],"messages":[{"id":"68e9077e6785a55bf6fe59bc731f26494b179eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 06:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85f83eea766adfae5b4b745c052e43b62b145220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 06:48:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3880b90921513bf2355eeff7e5e1ee770dece0e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 07:38:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e8604e4f451f3a6097e23ffd2760b6fd47bb1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 08:14:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7717d5b5a62a4861bc20fcb15c276a51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286e076a8251411780261abbe40e99ac : RETRY_LIMIT in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c7c06ec5c9148878c2d321ea4769295 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d8a30a4f5584cf1bc8c640c8329bc54 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fa056f0e1cb4368b2c20d28e5218951 : RETRY_LIMIT in 1m 46s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/a0343107f8984f628e19a7f0e6f97aca : SUCCESS in 11m 45s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/1db6c1ce1dd34b0b8d367e66fab9a534 : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a4244bbf3f9a3b2f6b8d53695b3d0766d584781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 08:27:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dea3deb7b26516c9d1e759233ed0660a831de85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:12:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d246c0bfa7424e2885ded282d2d16f03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d656d269bef04fc08fb8b8d3884d0c6f : RETRY_LIMIT in 1m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cde35ce07ba4f7f8455d99ee2bda10f : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56d33f4e54e44138bce1c75961b099b : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d71f41cba8394fd1956411384200cc80 : RETRY_LIMIT in 1m 34s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/b6ccf25a7298423390ecc6822f75887d : SUCCESS in 14m 23s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9880efd40bba4bc9950e29b760e54802 : SUCCESS in 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ecdbca7d53517d5af15c30b1a7bb9b9bfb17a8a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 10:14:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e48a3072a8e63ea43c4059ce0356b8a198236787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:00:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/235e72c2a89e43739743f6268d96b05a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8d05192f6449158e2c3a3d4891f57b : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74e575c444f6413f829c3b9df6e197a2 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49348f8bde264af3b421b3869754b6b0 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b19bcd51c6c4f7a9c8f34edd7b92bce : SUCCESS in 4m 00s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/0c5ab5a21d1541558be2b0431a25b7ff : SUCCESS in 12m 37s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/b874c126c0984715b1d8e94324cbf258 : FAILURE in 43m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70fdb1654d6d57cfcbb19d4b5abe3e6d772c84aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 11:48:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"49ef84cd6395d5fc778e3edf5830e3c8b7b946f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:45:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bb4496aef948f3873bb5e70a8d3a44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd354c04806a476eb1b1e91cc9330c25 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00b65272c04e4612976647d420b3aa46 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05e9bc6a69c64eefbbf1c5bd75df47e8 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ef1b86cd84b4e1d8e986670225966d9 : SUCCESS in 4m 18s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/81463368e42743c48b3f4a5af521326f : SUCCESS in 13m 13s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/91644b4e1e374840897b424ea7c73e64 : FAILURE in 51m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2e3143322ff82a02250563e547a57c3e019b72e7","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-04-16 12:53:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"12a19d9e876121a3ad9cbff98448047cb8e786f1","author":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"date":"2024-04-16 12:56:34.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"000c1943ee2f4c4ea32d60757aacd124fd5370c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:56:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"df898e23d53187dc53decd9162183ea7dec81920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 13:12:47.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45994aca0f8b473ca3a4e237564b9881\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe12c6e13b24feb846ad752af08ae64 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80ae7c52710645f4b633f30262c649ee : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e956ca88ae249ff92670db5424a16ef : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":6},{"id":"1eebc80237bec5c50e351434b75a1dd6da30c5f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 13:12:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7c42c56f1f28a2b5245dc0d89be923b59db28642","revisions":{"d99d568030fbe8fcc7b9dc73ce654154c7638da5":{"kind":"REWORK","_number":1,"created":"2024-04-16 06:23:48.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/1"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:27:24.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d99d568030fbe8fcc7b9dc73ce654154c7638da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/d99d568030fbe8fcc7b9dc73ce654154c7638da5"}]},"branch":"refs/heads/master"},"4c9058d1e48b95817cda395f1ff8728b59f13fbb":{"kind":"REWORK","_number":2,"created":"2024-04-16 06:48:53.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/2"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:52:20.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4c9058d1e48b95817cda395f1ff8728b59f13fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/4c9058d1e48b95817cda395f1ff8728b59f13fbb"}]},"branch":"refs/heads/master"},"a5ff998b9e97924ffbe747f35d0c8f1129c469ba":{"kind":"REWORK","_number":3,"created":"2024-04-16 07:38:26.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/3"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 07:41:55.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a5ff998b9e97924ffbe747f35d0c8f1129c469ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/a5ff998b9e97924ffbe747f35d0c8f1129c469ba"}]},"branch":"refs/heads/master"},"b86bb1b6c0cf27753d7a6cec85e8fe51c9bf4627":{"kind":"REWORK","_number":4,"created":"2024-04-16 08:27:26.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/4"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 08:30:57.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b86bb1b6c0cf27753d7a6cec85e8fe51c9bf4627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/b86bb1b6c0cf27753d7a6cec85e8fe51c9bf4627"}]},"branch":"refs/heads/master"},"e21d9eb12ab1bcf60d583e58b75a8962b4f98a3a":{"kind":"REWORK","_number":5,"created":"2024-04-16 10:14:20.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/5"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 10:17:50.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/e21d9eb12ab1bcf60d583e58b75a8962b4f98a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/e21d9eb12ab1bcf60d583e58b75a8962b4f98a3a"}]},"branch":"refs/heads/master"},"7c42c56f1f28a2b5245dc0d89be923b59db28642":{"kind":"REWORK","_number":6,"created":"2024-04-16 11:48:19.000000000","uploader":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"},"ref":"refs/changes/64/915964/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-agent","ref":"refs/changes/64/915964/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-agent refs/changes/64/915964/6"}}},"commit":{"parents":[{"commit":"29be6e96a6c590221e7d2fd1e30dffc7aec519a9","subject":"Update python testing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/29be6e96a6c590221e7d2fd1e30dffc7aec519a9"}]}],"author":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 06:26:47.000000000","tz":0},"committer":{"name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","date":"2024-04-16 11:51:48.000000000","tz":0},"subject":"fix: Making the gates happy","message":"fix: Making the gates happy\n\nChange-Id: Iddf3469f07486f2c58b46ea993547eecd6df3382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c42c56f1f28a2b5245dc0d89be923b59db28642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-agent/commit/7c42c56f1f28a2b5245dc0d89be923b59db28642"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35988,"name":"Hasan Acar","display_name":"Hasan Acar","email":"hasan.acar@tubitak.gov.tr","username":"hasanacar"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
