)]}'
{"id":"openstack%2Fmonasca-api~699958","triplet_id":"openstack%2Fmonasca-api~master~I2da1b3a60df4781c2e1bb954db3293722414c6b8","project":"openstack/monasca-api","branch":"master","hashtags":[],"change_id":"I2da1b3a60df4781c2e1bb954db3293722414c6b8","subject":"Fix invalid list of notification types in API spec","status":"MERGED","created":"2019-12-19 11:42:04.000000000","updated":"2020-01-16 08:25:41.000000000","submitted":"2020-01-16 08:24:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699958-1579163055586-d94f54cc","meta_rev_id":"6304171d00de85a4ea91d55a37adcdbd734e9b12","_number":699958,"virtual_id_number":699958,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 08:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"recommended":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":1,"date":"2020-01-16 03:40:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":1,"date":"2020-01-14 15:21:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-14 12:23:52.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 08:12:45.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 15:21:05.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2020-01-16 03:40:04.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2020-01-16 08:12:45.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-01-16 08:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b98034d71b75d79331dde192f84be5fc36e70a0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-19 11:42:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"733192aff6abc10f070705a991da8bf04f16841e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a54732c6da481faf34843f4b94dac9 : SUCCESS in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed4978f75bcb4823bcd6dbcc5d030e02 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2918a8cd4c754485afa1cedb2547725d : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9302a94aca46a6972f5094719d34f9 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d8b36a60b47491ea8daebc5fd80fb3a : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6035fef8c2d481086eec41670f0e1bd : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c811faba807cda3d5e6610411d11a84ea678f2a2","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-01-08 00:16:56.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good","accounts_in_message":[],"_revision_number":1},{"id":"0686dec436df1565d0e3be5e42b6562b19a42c91","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-14 09:54:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5987635a16432d568752520169b1fb48740f75b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-14 11:39:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"28e8d024659302353d05f76b0bb1595eb71dbe47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:01:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80c7a1c5d86b438cb09928832fd6f425 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c203cd49700f4e7885d958cf8dca77af : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff11c5e8b014d1faa8db65f2868e9dd : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42de1982c50746a4813dfd32ca23ce46 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/811096a780c745028489d8c40fe04e29 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9b98bf686a4bc89c6f76279952f86f : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"800f19214b76179eae38a3fcf5d301cab62696cd","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-14 12:23:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c28396247eed2c66103c0b7acfcf6bab89134a8c","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-01-14 15:21:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"14e4557e92198e37c075b6a087ecebf8e34c25b5","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2020-01-16 03:40:04.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f2aa364e44365fbbe3417795dd681812de51f467","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-01-16 08:12:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b96cccc44af074892147e726cc350f2da7735202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:12:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d94bf4c2fb904fadf535e9f05d942a27409e17f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:24:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"985d5beedb69759265be0703ce6ad4b196049d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:24:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9ba11175e046efa0a5de592b339669 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb130610bb5a482a96256162ee095dbc : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3819307209b474e9df330bbc4fc1319 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/989bb5bc65ef4cdd99c02d3bbfb5eb69 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67766525dfab4ad9ab9fdd757d472d3d : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"6304171d00de85a4ea91d55a37adcdbd734e9b12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 08:25:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d516bae1d84f49bdf0834099e80a57 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3a405648d37f85b03b9f4db064e717fa419a2661","revisions":{"8e93dffac031987b125a1b87125f61e4222b4c00":{"kind":"REWORK","_number":1,"created":"2019-12-19 11:42:04.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/58/699958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-api","ref":"refs/changes/58/699958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/1"}}},"commit":{"parents":[{"commit":"b3fb188fa7cdead7e44c8ff69148264033996120","subject":"Merge \"Update nodejs and npm packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/b3fb188fa7cdead7e44c8ff69148264033996120"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-19 11:27:37.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-19 11:41:51.000000000","tz":0},"subject":"Fix invalid list of notification types in API spec","message":"Fix invalid list of notification types in API spec\n\nTrivialFix\nChange-Id: I2da1b3a60df4781c2e1bb954db3293722414c6b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/8e93dffac031987b125a1b87125f61e4222b4c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/8e93dffac031987b125a1b87125f61e4222b4c00"}]},"branch":"refs/heads/master"},"3a405648d37f85b03b9f4db064e717fa419a2661":{"kind":"REWORK","_number":2,"created":"2020-01-14 11:39:44.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/58/699958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-api","ref":"refs/changes/58/699958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-api refs/changes/58/699958/2"}}},"commit":{"parents":[{"commit":"b3fb188fa7cdead7e44c8ff69148264033996120","subject":"Merge \"Update nodejs and npm packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/b3fb188fa7cdead7e44c8ff69148264033996120"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-19 11:27:37.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-01-14 11:39:24.000000000","tz":0},"subject":"Fix invalid list of notification types in API spec","message":"Fix invalid list of notification types in API spec\n\nTrivialFix\nChange-Id: I2da1b3a60df4781c2e1bb954db3293722414c6b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/3a405648d37f85b03b9f4db064e717fa419a2661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/3a405648d37f85b03b9f4db064e717fa419a2661"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
