)]}'
{"id":"openstack%2Fmonasca-api~731263","triplet_id":"openstack%2Fmonasca-api~master~I045d46d74809e506bcd94deaefc5477461616071","project":"openstack/monasca-api","branch":"master","topic":"stestr","hashtags":[],"change_id":"I045d46d74809e506bcd94deaefc5477461616071","subject":"Fix and unify cover environment","status":"MERGED","created":"2020-05-27 16:15:19.000000000","updated":"2020-05-28 15:42:38.000000000","submitted":"2020-05-28 15:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731263-1590680259931-60d2867d","meta_rev_id":"b167f3d1111281381041412b551dfab21ba9a94f","_number":731263,"virtual_id_number":731263,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 15:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"date":"2020-05-28 07:56:25.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 14:45:17.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":2,"date":"2020-05-28 11:37:31.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 14:45:17.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 11:37:31.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-28 14:45:17.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-05-28 15:37:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf8ecf909de16eee63f4d5321c34ec99f0002671","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-27 16:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2f104122ea1a570e278adb7078116b04c6a46dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 17:19:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/76036059b5fd4aa5be0d425799f08c10 : SUCCESS in 9m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d39b0664fa2248538f144f3811bc45eb : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/509b34a4f48c4edf8d4622c1648a781e : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757d7e6d65404b929c1b3b69422bb8c3 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a0152c0a154823beb38a13af73d3cf : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ff1ae44013041d299146aa941d857cd : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799a2562761045a5aa6ac0da6e4e6d13 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78b00b62b5094404b7781ef7c94b2a6d : SUCCESS in 4m 50s\n- monasca-tempest-log-python3-influxdb https://zuul.opendev.org/t/openstack/build/bbabdccf9eeb487587d5fd6008efd7f7 : SUCCESS in 46m 50s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8284aba7a6ba4c888e2b3f45f2a4c7dc : POST_FAILURE in 1h 00m 06s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/7328f84e3926460eb403755d9d2e9784 : FAILURE in 31m 37s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/a1aa7d5c9d2f4cab94f7d8a409531fc8 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"29b3e82160cdbe5687715b91d4e550c3b8a9683c","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-28 07:56:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b426aa7a7d6beddf30d3a794fb009255d889e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 08:48:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c9bca86ac29d42f9b1a8276c6424e6ec : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/485252b445e34988a2ec1dca68764e09 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de2de8de9e9e45b5be05bbb5e2c5d573 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1535f8210944eb851d059caceca1ac : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fa44e28ef3d4e04b8b44a2c11582be8 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/610c2b10566144a4b3a5d1a9c989497f : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4a7c96d9574047b68ac1c773e92a2c : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5620e717e0534d55ae403c1207bb73d8 : SUCCESS in 4m 23s\n- monasca-tempest-log-python3-influxdb https://zuul.opendev.org/t/openstack/build/360b0c759c224e809ae1fb3857226a3e : SUCCESS in 46m 34s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8fc7126d691e483b927bea5eff2fc4b2 : SUCCESS in 40m 48s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/9e201b2cbc36470aac8bd348c22717bc : FAILURE in 32m 39s (non-voting)\n- build-monasca-docker-image https://zuul.opendev.org/t/openstack/build/11b70a4860134bc6b9e2d25eb9e5cd74 : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9c228def9add32f6a923628e8bc4263e706861e8","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-28 11:37:31.000000000","message":"Patch Set 1: Code-Review+2\n\nWorks [1], nice, 74% coverage.\n\n[1] https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_485/731263/1/check/openstack-tox-cover/485252b/cover/","accounts_in_message":[],"_revision_number":1},{"id":"88f3340557b538a878b20884d40ef40016162ced","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-05-28 14:45:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"012b481bc492c6b2c5e51d428770e6d1010cd47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:45:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07f7c5fbbaabcd63ee8382317ca1fb17f4c4e326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:37:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4fa8210e5496464396cdb7977a973f4e : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4b8c6145f8046c3a5442305919ca331 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd24db4bc5ae47eabf2f656ff3a64d77 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93804861e994bd8b6aa3913a79ee83d : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29695f9f1c7b453bafda6fac95d07319 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb0f0e419fd4072adf55af3177e65da : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f90ccc941b8044cabae9d49d19fa90d1 : SUCCESS in 8m 12s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/9198ae187d6947ddb8755d0c0556df7a : SUCCESS in 44m 42s\n- monasca-tempest-log-python3-influxdb https://zuul.opendev.org/t/openstack/build/17c5cbe3b15a4a52a77cca48dc6910d9 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fe3357fe97a9c601eb959e8bdca9218d49d91b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:37:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b5d913ef776e7f8f426cd85e8904a7d9be455283","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:42:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7781b8d27b714bf18b549b150f5686f4 : SUCCESS in 3m 08s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143f6cdac86a448e8074e7da08a70212 : SUCCESS in 3m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06603abc0e59421b9ee2158ad8fcdcfd : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a99d2f063ae96e8a921ece4dbcd5601c6a827a25","revisions":{"a99d2f063ae96e8a921ece4dbcd5601c6a827a25":{"kind":"REWORK","_number":1,"created":"2020-05-27 16:15:19.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/63/731263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-api","ref":"refs/changes/63/731263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/63/731263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/63/731263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-api refs/changes/63/731263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-api refs/changes/63/731263/1"}}},"commit":{"parents":[{"commit":"738bd43559f2791dbc8f8b68f0be2b39c7a4e3be","subject":"Merge \"Move DB migrations documentation to its own file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/738bd43559f2791dbc8f8b68f0be2b39c7a4e3be"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-27 16:14:09.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-27 16:14:09.000000000","tz":120},"subject":"Fix and unify cover environment","message":"Fix and unify cover environment\n\nFix cover environment and unify it accordingly to Project Testing\nInterface [1].\n\n[1] https://governance.openstack.org/tc/reference/pti/python.html#coverage-jobs\n\nChange-Id: I045d46d74809e506bcd94deaefc5477461616071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/a99d2f063ae96e8a921ece4dbcd5601c6a827a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-api/commit/a99d2f063ae96e8a921ece4dbcd5601c6a827a25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
