)]}'
{"id":"openstack%2Fmonasca-ceilometer~639577","triplet_id":"openstack%2Fmonasca-ceilometer~master~If6eb225c74e417041a557426c796e814ffc40aba","project":"openstack/monasca-ceilometer","branch":"master","topic":"more-examples","hashtags":[],"change_id":"If6eb225c74e417041a557426c796e814ffc40aba","subject":"Adding example monasca_field_definition configuration files","status":"MERGED","created":"2019-02-27 06:55:32.000000000","updated":"2019-06-27 11:08:28.000000000","submitted":"2019-06-27 11:08:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639577-1561633708450-328d8c3f","meta_rev_id":"feb3eed39a151cdfe1393e32658f6a18e0cd7fa7","_number":639577,"virtual_id_number":639577,"owner":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-06-27 11:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"all":[{"value":2,"date":"2019-06-20 04:45:05.000000000","_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":1,"date":"2019-05-31 04:47:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 10:58:00.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-27 10:58:00.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 04:45:05.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-06-27 10:58:00.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-06-27 11:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"857157cc5f1f1845cff7719a955c7d638e56500e","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-02-27 06:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c560f7cacd6c990e72c474689da95afcbd9204ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 07:14:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/639577/1/check/openstack-tox-pep8/ad6c968/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/77/639577/1/check/openstack-tox-py27/ba9c81f/ : SUCCESS in 6m 52s\n- openstack-tox-py35 http://logs.openstack.org/77/639577/1/check/openstack-tox-py35/98049a8/ : SUCCESS in 8m 06s\n- openstack-tox-py36 http://logs.openstack.org/77/639577/1/check/openstack-tox-py36/bfa44ed/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/77/639577/1/check/openstack-tox-py37/e07b040/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"09be24fa8c8868215ad2ab19687d339c3270e48c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-02-27 08:45:17.000000000","message":"Patch Set 1:\n\nThanks Joseph, I\u0027d suggest creating new directory etc/ceilometer/examples and moving all yaml files in there. I like the idea with README in this directory.","accounts_in_message":[],"_revision_number":1},{"id":"bdccaa5676fa807cc8917ea5597cc991443cc5ad","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-31 00:38:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f91e099172eec3b0e967b86be215b2556cce9f4","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-31 00:44:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18624fa561a7d277c825f577d52c36e618ff8b6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 00:55:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/639577/3/check/openstack-tox-pep8/795dafe/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/77/639577/3/check/openstack-tox-py27/de7f7e0/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/77/639577/3/check/openstack-tox-py36/e836aaa/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/77/639577/3/check/openstack-tox-py37/fe84cab/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5bf4a3eda8735d900333c62f4bd14e7b6dc7ac30","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-31 04:47:06.000000000","message":"Patch Set 3: Code-Review+1\n\nI think this is ready to merge.  After merging, I may refactor a few of the other files and add more to the README in a separate patch.","accounts_in_message":[],"_revision_number":3},{"id":"bb6888c29bd799770e82f30790129cb47e1daff6","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-06-20 04:45:05.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM!","accounts_in_message":[],"_revision_number":3},{"id":"b35878196173bd5d53b551e5ec6fc5fe799d1e67","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-06-27 10:58:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"78aace360f7a24289c54772708cf8b5734bd6b0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:58:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3e3c50fcc8fbc35f7c0966627f90841e07c41e0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:08:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/639577/3/gate/openstack-tox-pep8/055e82f/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/77/639577/3/gate/openstack-tox-py27/a5a6d70/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/77/639577/3/gate/openstack-tox-py36/10e4ee4/ : SUCCESS in 4m 43s\n- openstack-tox-py37 http://logs.openstack.org/77/639577/3/gate/openstack-tox-py37/55b5e8e/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"29c32175224dceb12eba6c4fbacc055c5e62c881","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:08:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d43391d05a39e1b54eb922c4e204321c717e8ff0","revisions":{"52e1a0f6c0e5a26ab93d9b37c8d1f93f8da97075":{"kind":"REWORK","_number":1,"created":"2019-02-27 06:55:32.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/77/639577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/77/639577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/1"}}},"commit":{"parents":[{"commit":"93cd3ac9ecaeb7212aa42320aa6b087329ab07ee","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/93cd3ac9ecaeb7212aa42320aa6b087329ab07ee"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-02-27 06:53:58.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-02-27 06:55:13.000000000","tz":-480},"subject":"Adding example configuration files WIP","message":"Adding example configuration files WIP\n\nGathering up some configurations that have actually been used\nin production as examples for user configurations of Ceilosca.\n\nWIP - want to include a README that explains each set of configs\nWIP - need a storyboard\n\nChange-Id: If6eb225c74e417041a557426c796e814ffc40aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/52e1a0f6c0e5a26ab93d9b37c8d1f93f8da97075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/52e1a0f6c0e5a26ab93d9b37c8d1f93f8da97075"}]},"branch":"refs/heads/master"},"11b067eb184b981ce8a47f3add8481d57e2021cd":{"kind":"REWORK","_number":2,"created":"2019-05-31 00:38:49.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/77/639577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/77/639577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/2"}}},"commit":{"parents":[{"commit":"93cd3ac9ecaeb7212aa42320aa6b087329ab07ee","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/93cd3ac9ecaeb7212aa42320aa6b087329ab07ee"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-02-27 06:53:58.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-05-31 00:36:22.000000000","tz":-420},"subject":"Adding example monasca_field_definition configuration files","message":"Adding example monasca_field_definition configuration files\n\nGathering up some configurations that have actually been used\nin production as examples for user configurations of Ceilosca.\n\nStory: 2001766\nTask: 33543\n\nChange-Id: If6eb225c74e417041a557426c796e814ffc40aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/11b067eb184b981ce8a47f3add8481d57e2021cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/11b067eb184b981ce8a47f3add8481d57e2021cd"}]},"branch":"refs/heads/master"},"d43391d05a39e1b54eb922c4e204321c717e8ff0":{"kind":"REWORK","_number":3,"created":"2019-05-31 00:44:30.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/77/639577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/77/639577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/77/639577/3"}}},"commit":{"parents":[{"commit":"93cd3ac9ecaeb7212aa42320aa6b087329ab07ee","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/93cd3ac9ecaeb7212aa42320aa6b087329ab07ee"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-02-27 06:53:58.000000000","tz":-480},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-05-31 00:44:23.000000000","tz":-420},"subject":"Adding example monasca_field_definition configuration files","message":"Adding example monasca_field_definition configuration files\n\nGathering up some configurations that have actually been used\nin production as examples for user configurations of Ceilosca.\n\nStory: 2001766\nTask: 33543\n\nChange-Id: If6eb225c74e417041a557426c796e814ffc40aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/d43391d05a39e1b54eb922c4e204321c717e8ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/d43391d05a39e1b54eb922c4e204321c717e8ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
