)]}'
{"id":"openstack%2Fmonasca-ceilometer~687128","triplet_id":"openstack%2Fmonasca-ceilometer~master~I25b894e1ad222e3b55c6787c0625c03b9ebdcbf1","project":"openstack/monasca-ceilometer","branch":"master","topic":"readme-gnocchi-disable","hashtags":[],"change_id":"I25b894e1ad222e3b55c6787c0625c03b9ebdcbf1","subject":"Improve documentation on Ceilosca not needing Gnocchi","status":"MERGED","created":"2019-10-07 18:17:12.000000000","updated":"2019-10-22 21:22:33.000000000","submitted":"2019-10-22 09:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687128-1571737519168-2a96c748","meta_rev_id":"8b3a2701cabc9f5240987a0e3ee62c7de1975e61","_number":687128,"virtual_id_number":687128,"owner":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-15 07:26:13.000000000","_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 09:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 12:03:55.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-22 09:31:20.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 07:26:13.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2019-10-22 09:31:20.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-10-22 09:45:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c087c11e5882c9a6d9c8cedba924c6a90f7a5e5","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-07 18:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f61420fd13a7c284b96f22830a44b89589a877a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 18:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/beab8281605c4cbd859019354bef04be : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5c0c60fd4946e49be4366e91f72a48 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b47b23d11daf465eacb412b5be02139a : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c12e52cedc7145e1a8bd2989005be23b : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a069038254014dc7a9a84499e76cd663 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"498be2a1ada1ba3421d2f14ea1a438bd20ec5d8f","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-10-15 07:26:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fce7ae2d936c71d821648a215bcfb14562b1d90","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-15 14:38:32.000000000","message":"Patch Set 1: Review-Priority+1\n\nYes, I believe \u0027none\u0027 will work back through Ocata when Panko was split out.\nIt would not work in Newton (plugin.sh would die with an error).","accounts_in_message":[],"_revision_number":1},{"id":"bd2064e3a5230761951a9aae0e471ced26440bb9","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-18 12:03:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45f5d1524f5934b99833379f6dbe5b18dac18856","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-10-22 09:31:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc19b2c784a52f3324ad4a493a24d78b0e058b68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:31:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa9fa96d4db06f553144a42c8f8dcbf121425440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:45:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0941d5cabba84c8ea259ef29955f22e7 : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81eb8df605e445218449d8e142951479 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a468e20cdc4482287c73af6b45575ed : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4a3b1cc7664632b47eadd326e9a32b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ba6dfdb84e344809007f8f2b89526e9 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"55a8406467ace3ccfeb91196b9bb0c142c1e2040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 09:45:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d5d478a0e2aca37ead5b798d39131c7f1a16df8","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-10-22 21:22:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 94645c61f7ccacf0b83ecf6424e9b918db622cf1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"187bda14471e03d4ce2f989f2bb33d05dfc47e12","revisions":{"187bda14471e03d4ce2f989f2bb33d05dfc47e12":{"kind":"REWORK","_number":1,"created":"2019-10-07 18:17:12.000000000","uploader":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"ref":"refs/changes/28/687128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/28/687128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/28/687128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/28/687128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/28/687128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/28/687128/1"}}},"commit":{"parents":[{"commit":"15697d9d3af64a4d97656116bd7f03c14ded09b5","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/15697d9d3af64a4d97656116bd7f03c14ded09b5"}]}],"author":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-10-07 18:15:28.000000000","tz":-420},"committer":{"name":"Joseph Davis","email":"joseph.davis@suse.com","date":"2019-10-07 18:15:28.000000000","tz":-420},"subject":"Improve documentation on Ceilosca not needing Gnocchi","message":"Improve documentation on Ceilosca not needing Gnocchi\n\nThere was a recent question in IRC about how to configure Ceilosca.\nThese edits should make it clearer that Gnocchi is not needed\nin a Ceilosca deployment, and how to disable it.\n\nChange-Id: I25b894e1ad222e3b55c6787c0625c03b9ebdcbf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/187bda14471e03d4ce2f989f2bb33d05dfc47e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/187bda14471e03d4ce2f989f2bb33d05dfc47e12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
