)]}'
{"id":"openstack%2Fmonasca-ceilometer~715603","triplet_id":"openstack%2Fmonasca-ceilometer~master~Ica99c9a4affd2b693ef74f1e3ef7c8ea2c6a46d6","project":"openstack/monasca-ceilometer","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"Ica99c9a4affd2b693ef74f1e3ef7c8ea2c6a46d6","subject":"Update hacking for Python3","status":"MERGED","created":"2020-03-28 11:10:34.000000000","updated":"2020-04-15 21:48:43.000000000","submitted":"2020-04-15 21:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715603-1586987323198-cc994f1b","meta_rev_id":"6be2398fab9b7470ad6d3fdaffa40a4c61283c07","_number":715603,"virtual_id_number":715603,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 21:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":2,"date":"2020-04-15 21:37:42.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":1,"date":"2020-03-31 12:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-31 16:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":1,"date":"2020-04-15 21:37:42.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 12:51:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-31 16:59:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-15 21:35:49.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-04-15 21:37:42.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2020-04-15 21:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d34bb9709acf9506aaa85b168601cf81b7be768","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-28 11:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42f724ff7d1fc5731ea9800cfb451a06465534a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 11:17:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad96c71bb1b849bfaa55b5263429f38f : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85678a076fca489e9546c6c83919e928 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd7b74188f03489d9ca4f884ffedf063 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25c07792c571468495ee059ffd61f96f : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cee718e1f1bfeca497d55ab1520aef81e5bf189","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 10:32:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b5e2c5ca717e16b8529ab60816821c80be01ce49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 10:44:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e0ffbadb3e643b8b7fd2daf27578e15 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbdab4eb193a4c19aa1266ce5b010c9f : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b64d535ac9ee4c34a979484855c1b76c : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef52492baa9c4dbe8a336d24c999e7f5 : SUCCESS in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46a9b32882a047220e4d8f66341dfc85ce8888d2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-31 02:36:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"867df0c0553699059e046b92a4e6de65d44ea2ca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-31 06:49:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b5957d69278d835b6fdf22194e9daed42d63cf85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 07:03:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9429f65047b042c9b81cfe50923b349f : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43084d22e0564800b819a64f1203ade1 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caec06bbd2084411ab8fde698f5243c8 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c0ad811dbcd401993fe4747d677df35 : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49666de619f8b661701a1d5bf5a8aa5187d37ca4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 12:51:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"574445b2d8a2a2b2f4b1e285dbeea6999c80c544","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-31 16:59:39.000000000","message":"Patch Set 3: Code-Review+1\n\nthanks for updates.","accounts_in_message":[],"_revision_number":3},{"id":"36f8505253eef979353fe20ecce07d797cadd0f6","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2020-04-15 21:37:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nTested successfully, so merging.\n@Andreas: Note that we will be marking monasca-ceilometer as \u0027deprecated\u0027 soon, as the Monasca Publisher has been incorporated with Ceilometer.  But until then, it is good to have these kind of updates to keep it building correctly, so thank you.","accounts_in_message":[],"_revision_number":3},{"id":"99b903c9c32b7d36d3f43b90671004d2b11e745a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:37:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8b42d8fbbc89f2f9c4275a175a8698441b38a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:48:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f0d7a4b84040ad9de32e3f7311a3e8 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef09658c57d94e8e96b7494c0f2f0f42 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/187cf133c7c64f14a7318bf4103a052c : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ff041926f288c96b48257a547684db0339c5e60b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:48:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a0c6625d20f21384dbbaa4d0d7fa3bc4e127852b","revisions":{"d49f0e62fc6eabb9f4533678942203b32a9d23bf":{"kind":"REWORK","_number":1,"created":"2020-03-28 11:10:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/715603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/03/715603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/1"}}},"commit":{"parents":[{"commit":"4259f74ff1e1bbbd7f3a9524080be4819a226114","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/4259f74ff1e1bbbd7f3a9524080be4819a226114"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:06:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:10:23.000000000","tz":60},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: Ica99c9a4affd2b693ef74f1e3ef7c8ea2c6a46d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/d49f0e62fc6eabb9f4533678942203b32a9d23bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/d49f0e62fc6eabb9f4533678942203b32a9d23bf"}]},"branch":"refs/heads/master"},"690e4d7a61c37e925c9ba97d713b2e2ee8557c10":{"kind":"REWORK","_number":2,"created":"2020-03-30 10:32:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/715603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/03/715603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/2"}}},"commit":{"parents":[{"commit":"4259f74ff1e1bbbd7f3a9524080be4819a226114","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/4259f74ff1e1bbbd7f3a9524080be4819a226114"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:06:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-30 10:32:47.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: Ica99c9a4affd2b693ef74f1e3ef7c8ea2c6a46d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/690e4d7a61c37e925c9ba97d713b2e2ee8557c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/690e4d7a61c37e925c9ba97d713b2e2ee8557c10"}]},"branch":"refs/heads/master"},"a0c6625d20f21384dbbaa4d0d7fa3bc4e127852b":{"kind":"REWORK","_number":3,"created":"2020-03-31 06:49:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/03/715603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-ceilometer","ref":"refs/changes/03/715603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-ceilometer refs/changes/03/715603/3"}}},"commit":{"parents":[{"commit":"4259f74ff1e1bbbd7f3a9524080be4819a226114","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/4259f74ff1e1bbbd7f3a9524080be4819a226114"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 11:06:10.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-31 06:49:19.000000000","tz":0},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found.\n\nChange-Id: Ica99c9a4affd2b693ef74f1e3ef7c8ea2c6a46d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/a0c6625d20f21384dbbaa4d0d7fa3bc4e127852b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-ceilometer/commit/a0c6625d20f21384dbbaa4d0d7fa3bc4e127852b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
