)]}'
{"id":"openstack%2Fmonasca-common~603099","triplet_id":"openstack%2Fmonasca-common~master~Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd","project":"openstack/monasca-common","branch":"master","topic":"slack-patch","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-02-03 22:30:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd","subject":"Updates alarm to include value_meta field","status":"ABANDONED","created":"2018-09-17 11:18:49.000000000","updated":"2026-02-03 22:30:25.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6739d5c988b27922d7351f552ff663ad12912ce","_number":603099,"virtual_id_number":603099,"owner":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-22 08:42:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-17 16:55:30.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2018-09-18 13:21:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-24 12:41:47.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2026-02-03 22:30:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"d48038da6406d945a45b96a3e3a1d1ce237bce35","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 11:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caecd128de30a1bd19d6e17ff6e8854d825bdd3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-17 11:54:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/603099/1/check/openstack-tox-pep8/9b28354/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/99/603099/1/check/openstack-tox-py27/c852788/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/99/603099/1/check/openstack-tox-py35/0c725cd/ : SUCCESS in 3m 27s\n- legacy-monasca-common-maven-build http://logs.openstack.org/99/603099/1/check/legacy-monasca-common-maven-build/6d6e9a2/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/99/603099/1/check/openstack-tox-py36/35bd884/ : SUCCESS in 4m 06s\n- monasca-tempest-python-influxdb http://logs.openstack.org/99/603099/1/check/monasca-tempest-python-influxdb/bc6a07b/ : FAILURE in 17m 29s\n- monasca-tempest-java-influxdb http://logs.openstack.org/99/603099/1/check/monasca-tempest-java-influxdb/514409a/ : FAILURE in 20m 02s (non-voting)\n- monasca-tempest-python-cassandra http://logs.openstack.org/99/603099/1/check/monasca-tempest-python-cassandra/8e1dcf8/ : FAILURE in 20m 54s\n- monasca-tempest-java-cassandra http://logs.openstack.org/99/603099/1/check/monasca-tempest-java-cassandra/34c2ce4/ : FAILURE in 20m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/603099/1/check/openstack-tox-lower-constraints/6691ae6/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cdd0021e9c3fdb0abf31c9545361ac82856a021b","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 11:59:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8f2d93ecbea1bb3778c117c488198b91f7d76fe","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-09-17 12:16:24.000000000","message":"Patch Set 1:\n\n``````2018-09-17 11:53:07.233 | [ERROR] COMPILATION ERROR : \n2018-09-17 11:53:07.233 | [INFO] -------------------------------------------------------------\n2018-09-17 11:53:07.233 | [ERROR] /opt/stack/new/monasca-thresh/thresh/src/main/java/monasca/thresh/infrastructure/thresholding/AlarmThresholdingBolt.java:[235,9] no suitable constructor found for AlarmStateTransitionedEvent(java.lang.String,java.lang.String,java.lang.String,java.util.List\u003cmonasca.common.model.metric.MetricDefinition\u003e,java.lang.String,java.lang.String,monasca.common.model.alarm.AlarmState,monasca.common.model.alarm.AlarmState,java.lang.String,java.lang.String,java.lang.String,boolean,java.lang.String,java.util.List\u003cmonasca.common.model.alarm.AlarmTransitionSubAlarm\u003e,long)\n2018-09-17 11:53:07.233 |     constructor monasca.common.model.event.AlarmStateTransitionedEvent.AlarmStateTransitionedEvent() is not applicable\n2018-09-17 11:53:07.233 |       (actual and formal argument lists differ in length)\n2018-09-17 11:53:07.233 |     constructor monasca.common.model.event.AlarmStateTransitionedEvent.AlarmStateTransitionedEvent(java.lang.String,java.lang.String,java.lang.String,java.util.List\u003cmonasca.common.model.metric.MetricDefinition\u003e,java.lang.String,java.lang.String,monasca.common.model.alarm.AlarmState,monasca.common.model.alarm.AlarmState,java.lang.String,java.lang.String,java.lang.String,boolean,java.lang.String,java.lang.String,java.util.List\u003cmonasca.common.model.alarm.AlarmTransitionSubAlarm\u003e,long) is not applicable\n2018-09-17 11:53:07.233 |       (actual and formal argument lists differ in length)\n2018-09-17 11:53:07.233 | [INFO] 1 error\n2018-09-17 11:53:07.233 | [INFO] -------------------------------------------------------------\n2018-09-17 11:53:07.233 | [INFO] ------------------------------------------------------------------------\n2018-09-17 11:53:07.233 | [INFO] BUILD FAILURE\n2018-09-17 11:53:07.233 | [INFO] ------------------------------------------------------------------------\n2018-09-17 11:53:07.234 | [INFO] Total time: 10.751 s\n2018-09-17 11:53:07.234 | [INFO] Finished at: 2018-09-17T11:53:07+00:00\n2018-09-17 11:53:07.463 | [INFO] Final Memory: 48M/529M``````\n\n^ Looks like this is relevant to the test failure.","accounts_in_message":[],"_revision_number":1},{"id":"b75f3cb528cbb0684028814c406831d93bc4afbd","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 12:32:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ece1bd02e01bd4a2d855fe80aae63ba484f8849","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-09-17 12:57:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f69fd3bd00d4f17183dfd9c86124707b2a05a193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-17 13:15:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/603099/2/check/openstack-tox-pep8/a5305c3/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/99/603099/2/check/openstack-tox-py27/2ea725b/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/99/603099/2/check/openstack-tox-py35/19e2913/ : SUCCESS in 3m 25s\n- legacy-monasca-common-maven-build http://logs.openstack.org/99/603099/2/check/legacy-monasca-common-maven-build/cd6bcf9/ : FAILURE in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/99/603099/2/check/openstack-tox-py36/5f31ae7/ : SUCCESS in 4m 03s\n- monasca-tempest-python-influxdb http://logs.openstack.org/99/603099/2/check/monasca-tempest-python-influxdb/f2b7298/ : FAILURE in 18m 13s\n- monasca-tempest-java-influxdb http://logs.openstack.org/99/603099/2/check/monasca-tempest-java-influxdb/cef99ca/ : FAILURE in 19m 50s (non-voting)\n- monasca-tempest-python-cassandra http://logs.openstack.org/99/603099/2/check/monasca-tempest-python-cassandra/ee5469a/ : FAILURE in 19m 37s\n- monasca-tempest-java-cassandra http://logs.openstack.org/99/603099/2/check/monasca-tempest-java-cassandra/bd2b964/ : FAILURE in 22m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/603099/2/check/openstack-tox-lower-constraints/0cfd380/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"dc4861d8259b9d5088fdc4e04b516a45c58fae05","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 13:27:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a1626032f822c321c4d77a02acc546fa6e15f93","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 14:12:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3a53b23cbc891015687ec93edacc717fd60bbcf5","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 14:57:39.000000000","message":"Topic set to slack-patch","accounts_in_message":[],"_revision_number":4},{"id":"44e1515fa71b3798e3141c16244f0cf0a69e04e0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-09-17 15:16:05.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nOne nit to fix but looks good to me otherwise. Test coverage is via Devstack / Monasca Thresh right?","accounts_in_message":[],"_revision_number":4},{"id":"6727e130bd1df4a63adc350e523c3a00325f6a76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-17 15:17:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/603099/4/check/openstack-tox-pep8/06acec6/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/99/603099/4/check/openstack-tox-py27/ebcbee3/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/99/603099/4/check/openstack-tox-py35/5ff248d/ : SUCCESS in 3m 55s\n- legacy-monasca-common-maven-build http://logs.openstack.org/99/603099/4/check/legacy-monasca-common-maven-build/7a92274/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/99/603099/4/check/openstack-tox-py36/2e53e6f/ : SUCCESS in 3m 37s\n- monasca-tempest-python-influxdb http://logs.openstack.org/99/603099/4/check/monasca-tempest-python-influxdb/dcb218a/ : SUCCESS in 39m 13s\n- monasca-tempest-java-influxdb http://logs.openstack.org/99/603099/4/check/monasca-tempest-java-influxdb/6d8fe93/ : SUCCESS in 49m 29s (non-voting)\n- monasca-tempest-python-cassandra http://logs.openstack.org/99/603099/4/check/monasca-tempest-python-cassandra/25577ee/ : SUCCESS in 49m 04s\n- monasca-tempest-java-cassandra http://logs.openstack.org/99/603099/4/check/monasca-tempest-java-cassandra/b1c7a86/ : SUCCESS in 49m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/603099/4/check/openstack-tox-lower-constraints/65af8e5/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"4bed43837639cd56b828974cfb9b8c98a1ab2b76","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 15:43:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b1bd628e8bd597208b9450b63f2a4b2059900ccf","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-17 15:46:53.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n \u003e \n \u003e One nit to fix but looks good to me otherwise. Test coverage is via\n \u003e Devstack / Monasca Thresh right?\n\nYes, monasca-model AlarmStateTransitionedEvent class isn\u0027t tested in locally and only through tempest.","accounts_in_message":[],"_revision_number":5},{"id":"9969b2a58ef26472a4a6d97d0e7f72fd36c77914","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2018-09-17 16:55:30.000000000","message":"Patch Set 5: Code-Review+1\n\nLooks good. I\u0027ll wait for the bigger change to monasca-thresh to get further along before a +2.","accounts_in_message":[],"_revision_number":5},{"id":"3dc68b18e64b28584e288446520839e41c163005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-17 17:00:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/99/603099/5/check/openstack-tox-pep8/ebfd010/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/99/603099/5/check/openstack-tox-py27/3bd9aa7/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/99/603099/5/check/openstack-tox-py35/c21f660/ : SUCCESS in 4m 37s\n- legacy-monasca-common-maven-build http://logs.openstack.org/99/603099/5/check/legacy-monasca-common-maven-build/f977b98/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/99/603099/5/check/openstack-tox-py36/e8e77fb/ : SUCCESS in 3m 54s\n- monasca-tempest-python-influxdb http://logs.openstack.org/99/603099/5/check/monasca-tempest-python-influxdb/330b82b/ : SUCCESS in 51m 31s\n- monasca-tempest-java-influxdb http://logs.openstack.org/99/603099/5/check/monasca-tempest-java-influxdb/539f0d6/ : SUCCESS in 45m 27s (non-voting)\n- monasca-tempest-python-cassandra http://logs.openstack.org/99/603099/5/check/monasca-tempest-python-cassandra/880b327/ : SUCCESS in 49m 22s\n- monasca-tempest-java-cassandra http://logs.openstack.org/99/603099/5/check/monasca-tempest-java-cassandra/a404046/ : FAILURE in 28m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/603099/5/check/openstack-tox-lower-constraints/ea54054/ : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":5},{"id":"948c7ec8591b8694cde4c760e5a24ffbc9531c29","author":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"date":"2018-09-18 12:20:02.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0988fd4bcf0d8aff64517c1b3eace1c669e82cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-18 13:21:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/603099/5/check/openstack-tox-pep8/ef46e10/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/99/603099/5/check/openstack-tox-py27/8757ee1/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/99/603099/5/check/openstack-tox-py35/546486d/ : SUCCESS in 3m 55s\n- legacy-monasca-common-maven-build http://logs.openstack.org/99/603099/5/check/legacy-monasca-common-maven-build/89104e0/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/99/603099/5/check/openstack-tox-py36/197b535/ : SUCCESS in 4m 41s\n- monasca-tempest-python-influxdb http://logs.openstack.org/99/603099/5/check/monasca-tempest-python-influxdb/0115558/ : SUCCESS in 41m 57s\n- monasca-tempest-java-influxdb http://logs.openstack.org/99/603099/5/check/monasca-tempest-java-influxdb/ee55bbb/ : SUCCESS in 48m 57s (non-voting)\n- monasca-tempest-python-cassandra http://logs.openstack.org/99/603099/5/check/monasca-tempest-python-cassandra/009a4ff/ : SUCCESS in 56m 57s\n- monasca-tempest-java-cassandra http://logs.openstack.org/99/603099/5/check/monasca-tempest-java-cassandra/f57292a/ : SUCCESS in 44m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/603099/5/check/openstack-tox-lower-constraints/b7fe03f/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"adc4b88924b731f53611696eac560603c549756a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2018-09-24 12:41:47.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nLooks good. I\u0027m wondering if we could also add `currentValues` and `subAlarms` to this class. It would additionally make the notification more customizable.","accounts_in_message":[],"_revision_number":5},{"id":"9aa526f279a0b7475b711b63e2f5e669d3fc2b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 08:18:27.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1d00557370bf3f962f5148188be02ad6b86b72ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-22 08:18:35.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"80d8210d2c23e182951815e62c097360c4cc8eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 08:42:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e114f0859ab46798652820c94b85f28\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e5f4adb783f4b0a8259c0ab1299a01e : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53d21316f99417aac179bbf13a6d072 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/075ef9e1ec7d479d891b76db5586489f : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd233918c8474f41a42b583fb9ef0079 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2174129462ec4d2fb309219273266cbd : SUCCESS in 4m 48s (non-voting)\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/d54bf1645640491b9d420758624973c5 : FAILURE in 22m 33s (non-voting)\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/8b7f19b4c88f4eed94ce79ebb698abbf : FAILURE in 12m 57s (non-voting)\n- monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/1f0fc92836334034a673d7768e0697c3 : SUCCESS in 25s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/56f1337e891b408e983e54350f2edbb7 : SUCCESS in 1m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e6739d5c988b27922d7351f552ff663ad12912ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:30:25.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c150d1e02ab5ab7a97d80a77abdcc9ffa37d196c","revisions":{"bcd4e6593ee2d8b3439b072c68316c7f6522c89b":{"kind":"REWORK","_number":1,"created":"2018-09-17 11:18:49.000000000","uploader":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"ref":"refs/changes/99/603099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/1"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-17 11:18:47.000000000","tz":60},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nstory: #2001282\ntask: #5829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bcd4e6593ee2d8b3439b072c68316c7f6522c89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bcd4e6593ee2d8b3439b072c68316c7f6522c89b"}]},"branch":"refs/heads/master"},"8c0a7867d2473866bc2f6f5bc0c635384eb02885":{"kind":"REWORK","_number":2,"created":"2018-09-17 12:32:16.000000000","uploader":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"ref":"refs/changes/99/603099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/2"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-17 12:31:46.000000000","tz":60},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nstory: #2001282\ntask: #5829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/8c0a7867d2473866bc2f6f5bc0c635384eb02885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/8c0a7867d2473866bc2f6f5bc0c635384eb02885"}]},"branch":"refs/heads/master"},"779d128e57234076a7f04dc9f7d13356ddf72cde":{"kind":"REWORK","_number":3,"created":"2018-09-17 13:27:27.000000000","uploader":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"ref":"refs/changes/99/603099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/3"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Charana","email":"charananandasena@yahoo.com","date":"2018-09-17 13:26:50.000000000","tz":60},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nstory: #2001282\ntask: #5829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/779d128e57234076a7f04dc9f7d13356ddf72cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/779d128e57234076a7f04dc9f7d13356ddf72cde"}]},"branch":"refs/heads/master"},"27c09e96d18a372e76736d5fd6e840c93aa87ccd":{"kind":"NO_CHANGE","_number":4,"created":"2018-09-17 14:12:19.000000000","uploader":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"ref":"refs/changes/99/603099/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/4"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-17 14:12:04.000000000","tz":60},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nstory: #2001282\ntask: #5829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/27c09e96d18a372e76736d5fd6e840c93aa87ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/27c09e96d18a372e76736d5fd6e840c93aa87ccd"}]},"branch":"refs/heads/master"},"95bb7f73166756f72048d0f123757435c367cc79":{"kind":"REWORK","_number":5,"created":"2018-09-17 15:43:38.000000000","uploader":{"_account_id":28529,"name":"Charana Nandasena","email":"charana@stackhpc.com","username":"charana123"},"ref":"refs/changes/99/603099/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/5"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-17 15:43:07.000000000","tz":60},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nstory: #2001282\ntask: #5829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/95bb7f73166756f72048d0f123757435c367cc79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/95bb7f73166756f72048d0f123757435c367cc79"}]},"branch":"refs/heads/master"},"2e55fe2456a517c4b4f989cf7c91e05c10da3649":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-07-22 08:18:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/99/603099/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/6"}}},"commit":{"parents":[{"commit":"adfc935a62e48052972fd44380d4c80af3e5f753","subject":"Merge \"Remove not used pykafka requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/adfc935a62e48052972fd44380d4c80af3e5f753"}]}],"author":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 08:18:27.000000000","tz":0},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nstory: #2001282\ntask: #5829\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nSigned-off-by: Charana Nandasena \u003ccharana@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2e55fe2456a517c4b4f989cf7c91e05c10da3649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2e55fe2456a517c4b4f989cf7c91e05c10da3649"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c150d1e02ab5ab7a97d80a77abdcc9ffa37d196c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-22 08:18:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/99/603099/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/99/603099/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/99/603099/7"}}},"commit":{"parents":[{"commit":"bf623ef346852d39c6b6e8ec1449961ecf6cf165","subject":"Merge \"zuul: Move legacy jobs to project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bf623ef346852d39c6b6e8ec1449961ecf6cf165"}]}],"author":{"name":"Charana Nandasena","email":"charana@stackhpc.com","date":"2018-09-07 12:28:24.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-22 08:18:35.000000000","tz":0},"subject":"Updates alarm to include value_meta field","message":"Updates alarm to include value_meta field\n\nIncludes the value_meta field for the AlarmStateTransitionedEvent class\nst. it can be included in notifications.\n\nstory: #2001282\ntask: #5829\nChange-Id: Ibe57ade64dc791ba4ab5fd4b7df185d0de9195cd\nSigned-off-by: Charana Nandasena \u003ccharana@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/c150d1e02ab5ab7a97d80a77abdcc9ffa37d196c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/c150d1e02ab5ab7a97d80a77abdcc9ffa37d196c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
