)]}'
{"id":"openstack%2Fmonasca-common~637190","triplet_id":"openstack%2Fmonasca-common~master~I998dc4700db1826643b97e7e4a456c9bfcfe64fa","project":"openstack/monasca-common","branch":"master","topic":"metrics","hashtags":[],"change_id":"I998dc4700db1826643b97e7e4a456c9bfcfe64fa","subject":"Validate metric timestamp format","status":"ABANDONED","created":"2019-02-15 14:04:53.000000000","updated":"2026-02-03 22:30:46.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dada404e50ee67fcb1662a32fe45706e06a7c977","_number":637190,"virtual_id_number":637190,"owner":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-26 08:40:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-26 09:07:02.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-18 14:43:52.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-02-18 14:52:03.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-08-26 08:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-26 09:07:02.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2026-02-03 22:30:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"5c99b415c730718878fa7854e8ba1f51b676aa70","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-15 14:04:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1bf01686a9aceeb79bb5081263055cf8cb1f991","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-02-15 14:31:02.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nPlease undo all formatting changes and submit only the code that needs to be changed to implement the task.\n\nChanging the formatting makes it more difficult to see the actual logic that has changed and messes up the git history. For example, it makes git blame less effective.","accounts_in_message":[],"_revision_number":1},{"id":"770d8a8649046c640a9aa9c625d3f156ff109aa1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-02-15 14:31:06.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"beb02d9a8722719c244c6f39fb846c9c837b31f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 15:00:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/637190/1/check/openstack-tox-cover/fc4be00/cover/ : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/637190/1/check/openstack-tox-lower-constraints/5fa3e31/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/90/637190/1/check/openstack-tox-pep8/ff6eb3f/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/90/637190/1/check/openstack-tox-py27/3f4d378/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/90/637190/1/check/openstack-tox-py35/b679b8c/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/90/637190/1/check/openstack-tox-py36/71a8bdf/ : SUCCESS in 3m 39s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/90/637190/1/check/monasca-tempest-python2-influxdb/99e49e7/ : SUCCESS in 36m 28s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/90/637190/1/check/monasca-tempest-python3-influxdb/5995f84/ : SUCCESS in 38m 31s\n- monasca-tempest-java-influxdb http://logs.openstack.org/90/637190/1/check/monasca-tempest-java-influxdb/b16cf2e/ : FAILURE in 48m 10s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/90/637190/1/check/monasca-tempest-python2-cassandra/307baba/ : SUCCESS in 40m 07s\n- docker-build-monasca-base http://logs.openstack.org/90/637190/1/check/docker-build-monasca-base/3f3bcdb/ : SUCCESS in 2m 32s\n- legacy-monasca-common-maven-build http://logs.openstack.org/90/637190/1/check/legacy-monasca-common-maven-build/a1347c3/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1a2dc9ac064f8367f647644441338a5361240c4b","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-15 15:08:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f552268381a0a4405dda56e418714e25d8bbd54b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 16:08:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/637190/2/check/openstack-tox-cover/831f4ed/cover/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/637190/2/check/openstack-tox-lower-constraints/eabfd7e/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/90/637190/2/check/openstack-tox-pep8/c100f85/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/90/637190/2/check/openstack-tox-py27/5677b09/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/90/637190/2/check/openstack-tox-py35/3b15f8f/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/90/637190/2/check/openstack-tox-py36/5b7fb5e/ : SUCCESS in 3m 32s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/90/637190/2/check/monasca-tempest-python2-influxdb/36a2c26/ : SUCCESS in 40m 41s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/90/637190/2/check/monasca-tempest-python3-influxdb/600dcf0/ : SUCCESS in 37m 26s\n- monasca-tempest-java-influxdb http://logs.openstack.org/90/637190/2/check/monasca-tempest-java-influxdb/7222366/ : FAILURE in 50m 07s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/90/637190/2/check/monasca-tempest-python2-cassandra/2a1fa08/ : SUCCESS in 36m 17s\n- docker-build-monasca-base http://logs.openstack.org/90/637190/2/check/docker-build-monasca-base/4a66a54/ : SUCCESS in 3m 12s\n- legacy-monasca-common-maven-build http://logs.openstack.org/90/637190/2/check/legacy-monasca-common-maven-build/87f38b4/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"601168bdc0fb87971a7e3524a9af49b6c8d2f8cd","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-18 08:29:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da2f31421e4127ac76c7df58b8966ede901c3b31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 09:28:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/90/637190/3/check/openstack-tox-cover/e4899e3/ : FAILURE in 3m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/637190/3/check/openstack-tox-lower-constraints/763338e/ : FAILURE in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/90/637190/3/check/openstack-tox-pep8/c9447bc/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/90/637190/3/check/openstack-tox-py27/097bfe8/ : FAILURE in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/90/637190/3/check/openstack-tox-py35/f979acf/ : FAILURE in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/90/637190/3/check/openstack-tox-py36/2007236/ : FAILURE in 3m 57s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/90/637190/3/check/monasca-tempest-python2-influxdb/3e6197b/ : FAILURE in 29m 41s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/90/637190/3/check/monasca-tempest-python3-influxdb/edc8bcf/ : FAILURE in 27m 34s\n- monasca-tempest-java-influxdb http://logs.openstack.org/90/637190/3/check/monasca-tempest-java-influxdb/ba9d0c5/ : FAILURE in 49m 11s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/90/637190/3/check/monasca-tempest-python2-cassandra/48c5b32/ : FAILURE in 56m 38s\n- docker-build-monasca-base http://logs.openstack.org/90/637190/3/check/docker-build-monasca-base/46cb512/ : SUCCESS in 2m 44s\n- legacy-monasca-common-maven-build http://logs.openstack.org/90/637190/3/check/legacy-monasca-common-maven-build/7a87c0e/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5f582a5b45f5c5498527039e37bf6ff88a96f108","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-18 10:21:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5eb2404380edd2e605e1fce72e50d6b20d12bdf0","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-18 10:30:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"17242edde129a7951315b8841740dbe78b81da9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 11:22:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/90/637190/5/check/openstack-tox-cover/afffbd7/cover/ : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/637190/5/check/openstack-tox-lower-constraints/d22ea2d/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/90/637190/5/check/openstack-tox-pep8/01461fe/ : SUCCESS in 6m 20s\n- openstack-tox-py27 http://logs.openstack.org/90/637190/5/check/openstack-tox-py27/3be3345/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/90/637190/5/check/openstack-tox-py35/ca3a1c1/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/90/637190/5/check/openstack-tox-py36/7f0de1a/ : SUCCESS in 5m 05s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-python2-influxdb/ac0894d/ : FAILURE in 38m 07s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-python3-influxdb/37878b5/ : SUCCESS in 37m 30s\n- monasca-tempest-java-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-java-influxdb/bf8df35/ : FAILURE in 46m 06s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/90/637190/5/check/monasca-tempest-python2-cassandra/6c887f4/ : FAILURE in 40m 27s\n- docker-build-monasca-base http://logs.openstack.org/90/637190/5/check/docker-build-monasca-base/98d79e6/ : SUCCESS in 2m 24s\n- legacy-monasca-common-maven-build http://logs.openstack.org/90/637190/5/check/legacy-monasca-common-maven-build/a8ec4be/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"a9ed5d6b3a36d8aa900345974fd0a2eb492655cc","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-02-18 14:43:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c5e63af73a96d178b4d013b3654d889e53421393","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-02-18 14:52:03.000000000","message":"Patch Set 5:\n\nOne other thought on this: What if someone wants to load historical data into Monasca? I haven\u0027t looked at all the code paths that call this validation function: Will the metric be rejected at the API, or will it print a warning in the logs? A unit test would also be nice. Thanks for fixing the formatting changes.","accounts_in_message":[],"_revision_number":5},{"id":"f4672ec0894fb1282ecf21b514526bc765400c2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-18 15:42:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/637190/5/check/openstack-tox-cover/7962809/cover/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/637190/5/check/openstack-tox-lower-constraints/0741865/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/90/637190/5/check/openstack-tox-pep8/07cf641/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/90/637190/5/check/openstack-tox-py27/938cf0b/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/90/637190/5/check/openstack-tox-py35/c77faf2/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/90/637190/5/check/openstack-tox-py36/daa6668/ : SUCCESS in 6m 55s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-python2-influxdb/4f7fc07/ : SUCCESS in 30m 35s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-python3-influxdb/4bde4da/ : SUCCESS in 38m 54s\n- monasca-tempest-java-influxdb http://logs.openstack.org/90/637190/5/check/monasca-tempest-java-influxdb/3f258bc/ : FAILURE in 50m 14s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/90/637190/5/check/monasca-tempest-python2-cassandra/5de5322/ : SUCCESS in 36m 34s\n- docker-build-monasca-base http://logs.openstack.org/90/637190/5/check/docker-build-monasca-base/3dcef88/ : SUCCESS in 3m 06s\n- legacy-monasca-common-maven-build http://logs.openstack.org/90/637190/5/check/legacy-monasca-common-maven-build/aed4c8f/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":5},{"id":"2c7e00a8082194284c65198cb7e1ea7d78e494a5","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2019-02-22 14:56:51.000000000","message":"Patch Set 5:\n\nWhen timestamp is formatted \u0027158168661.2215465\u0027 (seconds in this way)  instead of \u00271581686612215.465\u0027 the following error occurs\n\n\n\nTraceback (most recent call last):                                                                          │\n│  File \"/opt/stack/monasca-persister/monasca_persister/repositories/persister.py\", line 56, in _flush       │\n│    self.repository.write_batch(self._data_points)                                                          │\n│  File \"/opt/stack/monasca-persister/monasca_persister/repositories/influxdb/abstract_repository.py\", line 3│\n│    self._influxdb_client.write_points(data_points, \u0027ms\u0027, protocol\u003d\u0027line\u0027)                                  │\n│  File \"/usr/local/lib/python3.6/dist-packages/influxdb/client.py\", line 483, in write_points               │\n│    tags\u003dtags, protocol\u003dprotocol)                                                                           │\n│  File \"/usr/local/lib/python3.6/dist-packages/influxdb/client.py\", line 544, in _write_points              │\n│    protocol\u003dprotocol                                                                                       │\n│  File \"/usr/local/lib/python3.6/dist-packages/influxdb/client.py\", line 327, in write                      │\n│    headers\u003dheaders                                                                                         │\n│  File \"/usr/local/lib/python3.6/dist-packages/influxdb/client.py\", line 286, in request                    │\n│    raise InfluxDBClientError(response.content, response.status_code)                                       │\n│influxdb.exceptions.InfluxDBClientError: 400: {\"error\":\"partial write: points beyond retention policy droppe│\n│                                                                                                            │\n│                                                                                                            │\n│During handling of the above exception, another exception occurred:                                         │\n│                                                                                                            │\n│Traceback (most recent call last):                                                                          │\n│  File \"/opt/stack/monasca-persister/monasca_persister/repositories/persister.py\", line 85, in run          │\n│    self._flush()                                                                                           │\n│  File \"/opt/stack/monasca-persister/monasca_persister/repositories/persister.py\", line 64, in _flush       │\n│    if \"partial write: points beyond retention policy dropped\" in ex.message:                               │\n│AttributeError: \u0027InfluxDBClientError\u0027 object has no attribute \u0027message\u0027","accounts_in_message":[],"_revision_number":5},{"id":"0578de2d8cb6fa5ea1b65f4216094078fa4ad2e9","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2020-08-26 07:46:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"639f9e1bf201c6f8324b248e94a8d07bc204d2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:40:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebd70f8412344471baac3d2cd3196e82 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4565a794839a44a896cdfb676ca13e0f : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2715313854740b6b19668a00c2687cb : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0afc9667c28944668572c1c16cf6e9c2 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d584b3bc4a4c9189a49bb0da376e27 : SUCCESS in 5m 15s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/226e5cbc5cd342b4bacd36e85b5fdda9 : SUCCESS in 43m 32s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/87e68cbaabc84c73a5320f6cbe535b47 : FAILURE in 32m 21s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/8c1140d781e6411292c15067734a778d : SUCCESS in 2m 45s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/09bb885eaa4540ac87dba9540221bb38 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"2bf03f83829af1566fe72164b1b04450a3030f8e","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-08-26 09:07:02.000000000","message":"Patch Set 6:\n\nI understand Doug\u0027s concern. Current implementation will not allow pushing historical data from 2001 and older. While not really relevant for monitoring, it could be a problem for scientific applications.\n\nThe referenced exception thrown later is a matter of TSDB configuration.\n\nHow about introducing a flag in monasca-api configuration like e.g. `allow_historical` with default `false`?","accounts_in_message":[],"_revision_number":6},{"id":"dada404e50ee67fcb1662a32fe45706e06a7c977","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-02-03 22:30:46.000000000","message":"Abandoned\n\nThis project has been retired and no longer accepts code changes via https://review.opendev.org. New changes can be uploaded if the project is restored.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"701dfc4f6c394d5206859ab2240597a08e0cc956","revisions":{"ab4e9681deabddcef37c79090890af30ef850f3f":{"kind":"REWORK","_number":1,"created":"2019-02-15 14:04:53.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/1"}}},"commit":{"parents":[{"commit":"5946f650b9be34fdcce0c71ded66cc4727df03dd","subject":"Merge \"Add tempest tests jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/5946f650b9be34fdcce0c71ded66cc4727df03dd"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:04:51.000000000","tz":60},"subject":"Validate metric timestamp format","message":"Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ab4e9681deabddcef37c79090890af30ef850f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ab4e9681deabddcef37c79090890af30ef850f3f"}]},"branch":"refs/heads/master"},"12ccd9f9474a750a941330b17d914359077c2df5":{"kind":"REWORK","_number":2,"created":"2019-02-15 15:08:57.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/2"}}},"commit":{"parents":[{"commit":"5946f650b9be34fdcce0c71ded66cc4727df03dd","subject":"Merge \"Add tempest tests jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/5946f650b9be34fdcce0c71ded66cc4727df03dd"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 15:08:36.000000000","tz":60},"subject":"[WiP] Validate metric timestamp format","message":"[WiP] Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/12ccd9f9474a750a941330b17d914359077c2df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/12ccd9f9474a750a941330b17d914359077c2df5"}]},"branch":"refs/heads/master"},"db237bbff837e486bda3585d56eb3bbdf9ff05e3":{"kind":"REWORK","_number":3,"created":"2019-02-18 08:29:33.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/3"}}},"commit":{"parents":[{"commit":"5946f650b9be34fdcce0c71ded66cc4727df03dd","subject":"Merge \"Add tempest tests jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/5946f650b9be34fdcce0c71ded66cc4727df03dd"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-18 08:29:16.000000000","tz":60},"subject":"Validate metric timestamp format","message":"Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/db237bbff837e486bda3585d56eb3bbdf9ff05e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/db237bbff837e486bda3585d56eb3bbdf9ff05e3"}]},"branch":"refs/heads/master"},"7f8638ac9dcd9f5477ff63c0568e15e7992223d9":{"kind":"REWORK","_number":4,"created":"2019-02-18 10:21:28.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/4"}}},"commit":{"parents":[{"commit":"5946f650b9be34fdcce0c71ded66cc4727df03dd","subject":"Merge \"Add tempest tests jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/5946f650b9be34fdcce0c71ded66cc4727df03dd"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-18 10:21:16.000000000","tz":60},"subject":"Validate metric timestamp format","message":"Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/7f8638ac9dcd9f5477ff63c0568e15e7992223d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/7f8638ac9dcd9f5477ff63c0568e15e7992223d9"}]},"branch":"refs/heads/master"},"1468e8e21bc8102b57df3cb61efbf4527f75e33d":{"kind":"REWORK","_number":5,"created":"2019-02-18 10:30:41.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/5"}}},"commit":{"parents":[{"commit":"5946f650b9be34fdcce0c71ded66cc4727df03dd","subject":"Merge \"Add tempest tests jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/5946f650b9be34fdcce0c71ded66cc4727df03dd"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-18 10:30:32.000000000","tz":60},"subject":"Validate metric timestamp format","message":"Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1468e8e21bc8102b57df3cb61efbf4527f75e33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1468e8e21bc8102b57df3cb61efbf4527f75e33d"}]},"branch":"refs/heads/master"},"701dfc4f6c394d5206859ab2240597a08e0cc956":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-26 07:46:36.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/90/637190/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/90/637190/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/90/637190/6"}}},"commit":{"parents":[{"commit":"3a7030d1684c3f3ca98007818db62e5543a3381f","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3a7030d1684c3f3ca98007818db62e5543a3381f"}]}],"author":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2019-02-15 14:03:13.000000000","tz":60},"committer":{"name":"Michał Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2020-08-26 07:46:36.000000000","tz":0},"subject":"Validate metric timestamp format","message":"Validate metric timestamp format\n\nMonasca should validate timestamp format. Metrics send via Monasca Agent\nare properly formatted but It is possible to send metrics to API with\ntimestamp formatted to seconds, not milliseconds.\n\nChange-Id: I998dc4700db1826643b97e7e4a456c9bfcfe64fa\nStory: 2004998\nTask: 29470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/701dfc4f6c394d5206859ab2240597a08e0cc956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/701dfc4f6c394d5206859ab2240597a08e0cc956"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
