)]}'
{"id":"openstack%2Fmonasca-common~652418","triplet_id":"openstack%2Fmonasca-common~master~If9c46ca5a80128bb8c6a9bd9863c4c3305bae54c","project":"openstack/monasca-common","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"If9c46ca5a80128bb8c6a9bd9863c4c3305bae54c","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-14 23:54:28.000000000","updated":"2019-05-16 09:46:37.000000000","submitted":"2019-05-16 09:46:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652418-1557999997461-96febb6b","meta_rev_id":"49b8ef09d410d3b257b2541938fcfab4ccb27e23","_number":652418,"virtual_id_number":652418,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":2,"date":"2019-05-16 09:46:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"recommended":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"all":[{"value":1,"date":"2019-05-16 09:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":1,"date":"2019-04-23 07:27:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":2,"date":"2019-05-13 22:52:29.000000000","_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 11:01:55.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":1,"date":"2019-05-14 08:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 11:01:55.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 07:27:34.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-05-13 22:52:29.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-05-14 08:57:41.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-14 11:01:55.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-05-16 09:16:44.000000000","updated_by":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"reviewer":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"state":"REVIEWER"},{"updated":"2019-05-16 09:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"768b175af61415d336cc50a3b3cfbf2b5b788665","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-14 23:54:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"856f630b3970853bdb892839cd30644d6f544cce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 00:41:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/1/check/test-release-openstack/c5f30f4/ : SUCCESS in 4m 52s\n- openstack-tox-cover http://logs.openstack.org/18/652418/1/check/openstack-tox-cover/d342a5d/ : FAILURE in 3m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/1/check/openstack-tox-lower-constraints/09231a9/ : FAILURE in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/1/check/openstack-tox-pep8/b310d00/ : FAILURE in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/1/check/openstack-tox-py27/6f93bac/ : FAILURE in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/1/check/openstack-tox-py36/f302b5f/ : FAILURE in 5m 15s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/1/check/monasca-tempest-python2-influxdb/7405874/ : SUCCESS in 34m 47s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/1/check/monasca-tempest-python3-influxdb/a05c1aa/ : SUCCESS in 36m 49s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/1/check/monasca-tempest-java-influxdb/84147a1/ : FAILURE in 45m 09s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/1/check/monasca-tempest-python2-cassandra/1b13626/ : SUCCESS in 37m 58s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/1/check/build-monasca-common-docker-base-image/9093788/ : SUCCESS in 2m 06s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/1/check/legacy-monasca-common-maven-build/9cab8d6/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5195fe098436adcb05d82603159ca4d9628f56b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 18:14:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d231bb10cfd77c371f587c463d39d25978adfd0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 19:41:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/2/check/test-release-openstack/2076897/ : SUCCESS in 3m 12s\n- openstack-tox-cover http://logs.openstack.org/18/652418/2/check/openstack-tox-cover/565a9d2/ : FAILURE in 3m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/2/check/openstack-tox-lower-constraints/7f2c262/ : FAILURE in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/2/check/openstack-tox-pep8/b4b9833/ : FAILURE in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/2/check/openstack-tox-py27/ab02130/ : FAILURE in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/2/check/openstack-tox-py36/fe8c851/ : FAILURE in 4m 07s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/2/check/monasca-tempest-python2-influxdb/0edfc49/ : SUCCESS in 34m 26s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/2/check/monasca-tempest-python3-influxdb/5466218/ : SUCCESS in 32m 42s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/2/check/monasca-tempest-java-influxdb/eda981c/ : FAILURE in 46m 16s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/2/check/monasca-tempest-python2-cassandra/2ad4e58/ : SUCCESS in 42m 55s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/2/check/build-monasca-common-docker-base-image/bfe27e4/ : SUCCESS in 2m 32s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/2/check/legacy-monasca-common-maven-build/daf7966/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"63945080bbe7cbe5cfa72d5d1ff2deabc152c520","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-04-15 20:08:57.000000000","message":"Patch Set 2: Code-Review+1\n\nlooks good, but need to look in to why zuul disagrees.","accounts_in_message":[],"_revision_number":2},{"id":"a9120be7b7511e4b92d5aff44b9e46bd929ee1cc","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-04-17 04:37:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf11b507de592bf10bb4aae6ad3be09793c35dd4","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-04-17 04:39:18.000000000","message":"Removed Code-Review+1 by Akhil jain \u003cakhil.jain@india.nec.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"6e46c80c93e06e6e797e9320431091ced014179b","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-04-17 05:08:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\none comment inline, I think that will fix the issue.","accounts_in_message":[],"_revision_number":2},{"id":"66c0200cf79fd3a97502b4edc55c4b2eff176b0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-23 04:32:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fc7726cf58effe2133d9dd73fc51181e06ca3883","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 05:33:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/check/test-release-openstack/e6eb791/ : SUCCESS in 3m 32s\n- openstack-tox-cover http://logs.openstack.org/18/652418/3/check/openstack-tox-cover/e87b36b/cover/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/check/openstack-tox-lower-constraints/4c9f859/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/check/openstack-tox-pep8/f607d11/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/check/openstack-tox-py27/19e52ef/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/check/openstack-tox-py36/d65cc7d/ : SUCCESS in 3m 06s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-influxdb/f6b9a4f/ : SUCCESS in 40m 03s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python3-influxdb/1edae71/ : SUCCESS in 43m 15s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-java-influxdb/74f1619/ : FAILURE in 44m 03s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-cassandra/235e95f/ : SUCCESS in 39m 18s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/3/check/build-monasca-common-docker-base-image/3810c53/ : SUCCESS in 2m 07s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/check/legacy-monasca-common-maven-build/5b8aad3/ : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"99c556b51fc6d8cb43119ed0139a89a1b5582fcd","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-04-23 07:27:34.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":3},{"id":"4d86bf6d91d00f6722d81ef31e2ce3d6ea52e099","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-13 22:52:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"310159c8058f779a7aefe80fe6b93d50282ac74d","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 08:57:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e8aa920df78d5d3aaed62d8232cc3594122db6cf","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-14 11:01:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a4b656d169e81b0edf876767b7dd45e20568742","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:02:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b2a0601a8cc06fa3fbbbd4b9a38e6022ec6dbb0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:37:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/gate/test-release-openstack/5247f3f/ : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/gate/openstack-tox-lower-constraints/5dcf611/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/gate/openstack-tox-pep8/b95f2d9/ : FAILURE in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/gate/openstack-tox-py27/2c67fd6/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/gate/openstack-tox-py36/b74ad4f/ : SUCCESS in 3m 38s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python2-influxdb/3fcdc86/ : SUCCESS in 31m 52s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python3-influxdb/3758071/ : SUCCESS in 32m 35s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python2-cassandra/3fca9e6/ : SUCCESS in 34m 13s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/gate/legacy-monasca-common-maven-build/95beac5/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"8b3c38ce51259ebabd5dd9a804d0eb40f7c8747d","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-14 14:49:38.000000000","message":"Patch Set 3:\n\nWill wait for the tox/bandit fix to land before rechecking this.","accounts_in_message":[],"_revision_number":3},{"id":"0a8045434a93452946f51125be21c1b08a011f6d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-15 11:03:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e51a247ecc00cd477a85b0e9ffa48b4f7b373a2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 11:52:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/check/test-release-openstack/8348ebe/ : SUCCESS in 2m 49s\n- openstack-tox-cover http://logs.openstack.org/18/652418/3/check/openstack-tox-cover/e9b59ee/cover/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/check/openstack-tox-lower-constraints/43e3a56/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/check/openstack-tox-pep8/4e20670/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/check/openstack-tox-py27/03b4957/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/check/openstack-tox-py36/863619d/ : SUCCESS in 4m 11s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-influxdb/e0e1fd1/ : FAILURE in 36m 18s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python3-influxdb/c78f457/ : FAILURE in 39m 47s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-java-influxdb/faf097d/ : FAILURE in 19m 00s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-cassandra/552f9cb/ : SUCCESS in 41m 40s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/3/check/build-monasca-common-docker-base-image/1146b93/ : SUCCESS in 2m 06s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/check/legacy-monasca-common-maven-build/13fd97b/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"1a4d6a2b00cf91f8a0521a3065ce45e46ca9129f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-15 13:50:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"55f2c932b7a5ae8b985ad0aa83555e4defc38c19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:42:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/check/test-release-openstack/805d26e/ : SUCCESS in 2m 10s\n- openstack-tox-cover http://logs.openstack.org/18/652418/3/check/openstack-tox-cover/d83fe8e/cover/ : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/check/openstack-tox-lower-constraints/3744e23/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/check/openstack-tox-pep8/1291d23/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/check/openstack-tox-py27/24873dd/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/check/openstack-tox-py36/07113ae/ : SUCCESS in 3m 52s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-influxdb/21cd884/ : SUCCESS in 42m 21s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python3-influxdb/c906a7f/ : SUCCESS in 39m 33s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-java-influxdb/66f6009/ : FAILURE in 47m 52s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-cassandra/f3a028f/ : FAILURE in 36m 49s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/3/check/build-monasca-common-docker-base-image/16ca671/ : SUCCESS in 2m 28s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/check/legacy-monasca-common-maven-build/4263f15/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7722f1182b6eac1583f407d9c54e13ab8cfef0ea","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-15 22:06:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce88fd08c4b7c7fd3500eb08059022ca7b8407a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:53:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/check/test-release-openstack/4984b52/ : SUCCESS in 2m 09s\n- openstack-tox-cover http://logs.openstack.org/18/652418/3/check/openstack-tox-cover/1e2bec1/cover/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/check/openstack-tox-lower-constraints/55bd502/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/check/openstack-tox-pep8/26b89ad/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/check/openstack-tox-py27/2ef0151/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/check/openstack-tox-py36/06fa254/ : SUCCESS in 3m 13s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-influxdb/9f0bc30/ : SUCCESS in 38m 09s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python3-influxdb/1df575f/ : FAILURE in 14m 56s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-java-influxdb/d2a7b68/ : FAILURE in 44m 29s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-cassandra/261a27b/ : SUCCESS in 36m 01s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/3/check/build-monasca-common-docker-base-image/0c5e98d/ : SUCCESS in 1m 48s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/check/legacy-monasca-common-maven-build/f0ba68d/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ad2c1301f07ebb5947bcca78441ccfc4999b9ca1","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-05-16 07:49:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb0ed52395fbe36dcdb7339c8528f830694f0be7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:03:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/check/test-release-openstack/fa57939/ : SUCCESS in 2m 37s\n- openstack-tox-cover http://logs.openstack.org/18/652418/3/check/openstack-tox-cover/bd9dfc9/cover/ : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/check/openstack-tox-lower-constraints/2b06384/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/check/openstack-tox-pep8/fae7f76/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/check/openstack-tox-py27/5302332/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/check/openstack-tox-py36/84c88fc/ : SUCCESS in 3m 25s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-influxdb/a856a87/ : SUCCESS in 38m 34s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-python3-influxdb/82edf68/ : SUCCESS in 40m 16s\n- monasca-tempest-java-influxdb http://logs.openstack.org/18/652418/3/check/monasca-tempest-java-influxdb/4700aa1/ : FAILURE in 49m 48s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/check/monasca-tempest-python2-cassandra/fcb59d6/ : SUCCESS in 39m 05s\n- build-monasca-common-docker-base-image http://logs.openstack.org/18/652418/3/check/build-monasca-common-docker-base-image/c477e60/ : SUCCESS in 2m 07s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/check/legacy-monasca-common-maven-build/83aa932/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"9b5ea0218789a624adcb5c7b109e144740bcc980","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:03:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b8beb90877ce89ebf4e9376cd012074f7564a84","author":{"_account_id":30448,"name":"moser-cui","email":"cuilei@awcloud.com","username":"cuilei"},"date":"2019-05-16 09:16:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d7d7845500a7b2c60762859f5e9c4bb326730b70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:46:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5790f0214ee581c4b3cacc74b1f45205b3190e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:46:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/18/652418/3/gate/test-release-openstack/80ea353/ : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/652418/3/gate/openstack-tox-lower-constraints/06d408d/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/18/652418/3/gate/openstack-tox-pep8/6974fe3/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/18/652418/3/gate/openstack-tox-py27/429922b/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/18/652418/3/gate/openstack-tox-py36/2fb4926/ : SUCCESS in 3m 33s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python2-influxdb/e002cec/ : SUCCESS in 34m 14s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python3-influxdb/1c85102/ : SUCCESS in 35m 45s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/18/652418/3/gate/monasca-tempest-python2-cassandra/18724a4/ : SUCCESS in 40m 08s\n- legacy-monasca-common-maven-build http://logs.openstack.org/18/652418/3/gate/legacy-monasca-common-maven-build/0083dd3/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"625ed074917da571bb3c85d2bd5729c47b887bab","revisions":{"39f55b74adaba06f698de5a05fee5f25c9672d5d":{"kind":"REWORK","_number":1,"created":"2019-04-14 23:54:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/652418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/18/652418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/1"}}},"commit":{"parents":[{"commit":"bda421c4c636953ceebc63a2ba065543546aa12c","subject":"Merge \"Use proper naming for docker services image zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bda421c4c636953ceebc63a2ba065543546aa12c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:54:09.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:54:26.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: If9c46ca5a80128bb8c6a9bd9863c4c3305bae54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/39f55b74adaba06f698de5a05fee5f25c9672d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/39f55b74adaba06f698de5a05fee5f25c9672d5d"}]},"branch":"refs/heads/master"},"627840c023275923abd79791acf33966c8b6d295":{"kind":"REWORK","_number":2,"created":"2019-04-15 18:14:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/652418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/18/652418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/2"}}},"commit":{"parents":[{"commit":"bda421c4c636953ceebc63a2ba065543546aa12c","subject":"Merge \"Use proper naming for docker services image zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bda421c4c636953ceebc63a2ba065543546aa12c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:54:09.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 18:14:55.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: If9c46ca5a80128bb8c6a9bd9863c4c3305bae54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/627840c023275923abd79791acf33966c8b6d295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/627840c023275923abd79791acf33966c8b6d295"}]},"branch":"refs/heads/master"},"625ed074917da571bb3c85d2bd5729c47b887bab":{"kind":"REWORK","_number":3,"created":"2019-04-23 04:32:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/18/652418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/18/652418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/18/652418/3"}}},"commit":{"parents":[{"commit":"bda421c4c636953ceebc63a2ba065543546aa12c","subject":"Merge \"Use proper naming for docker services image zuul jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bda421c4c636953ceebc63a2ba065543546aa12c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-14 23:54:09.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-23 04:32:17.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: If9c46ca5a80128bb8c6a9bd9863c4c3305bae54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/625ed074917da571bb3c85d2bd5729c47b887bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/625ed074917da571bb3c85d2bd5729c47b887bab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
