)]}'
{"id":"openstack%2Fmonasca-common~659093","triplet_id":"openstack%2Fmonasca-common~master~If2e5f82279981ced89da6fc134a238105be60d8b","project":"openstack/monasca-common","branch":"master","hashtags":[],"change_id":"If2e5f82279981ced89da6fc134a238105be60d8b","subject":"Blacklist bandit 1.6.0","status":"MERGED","created":"2019-05-14 15:00:35.000000000","updated":"2019-05-15 11:01:33.000000000","submitted":"2019-05-15 11:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659093-1557918093978-881d6cdd","meta_rev_id":"12126976738c311490f20cc87e2327380bfada79","_number":659093,"virtual_id_number":659093,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-05-15 11:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-15 09:19:57.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":2,"date":"2019-05-15 07:16:11.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":1,"date":"2019-05-15 07:16:11.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 07:16:11.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-05-15 11:01:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0d33f7f5af20cfd03fcdaa34ce54a2a50d51c11","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-14 15:00:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3978a6e58ecf1cc987606b20941dad6df67b5d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 15:47:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/93/659093/1/check/requirements-check/89fd285/ : SUCCESS in 2m 20s\n- openstack-tox-cover http://logs.openstack.org/93/659093/1/check/openstack-tox-cover/cab000a/cover/ : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/659093/1/check/openstack-tox-lower-constraints/4cc614b/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/93/659093/1/check/openstack-tox-pep8/918b5f3/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/93/659093/1/check/openstack-tox-py27/8d1f516/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/93/659093/1/check/openstack-tox-py35/ca4d723/ : SUCCESS in 7m 13s\n- openstack-tox-py36 http://logs.openstack.org/93/659093/1/check/openstack-tox-py36/0d59104/ : SUCCESS in 3m 20s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-influxdb/4cec642/ : FAILURE in 15m 19s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python3-influxdb/2490514/ : SUCCESS in 41m 51s\n- monasca-tempest-java-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-java-influxdb/82b11d9/ : FAILURE in 46m 31s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-cassandra/a0929fb/ : SUCCESS in 38m 12s\n- build-monasca-common-docker-base-image http://logs.openstack.org/93/659093/1/check/build-monasca-common-docker-base-image/bcbe8b6/ : SUCCESS in 2m 39s\n- legacy-monasca-common-maven-build http://logs.openstack.org/93/659093/1/check/legacy-monasca-common-maven-build/32ed4d8/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3e91ca715cbc8791809e170f17883c6951872602","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-15 07:16:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0df77d67a545dae7686e45f0260cc6c4dbc4ac2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 08:05:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/93/659093/1/check/requirements-check/aa822db/ : SUCCESS in 2m 41s\n- openstack-tox-cover http://logs.openstack.org/93/659093/1/check/openstack-tox-cover/49b6be8/cover/ : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/659093/1/check/openstack-tox-lower-constraints/cec77a1/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/93/659093/1/check/openstack-tox-pep8/6514b46/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/93/659093/1/check/openstack-tox-py27/2fdf4d1/ : SUCCESS in 6m 26s\n- openstack-tox-py35 http://logs.openstack.org/93/659093/1/check/openstack-tox-py35/9fa553f/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/93/659093/1/check/openstack-tox-py36/44af097/ : SUCCESS in 3m 10s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-influxdb/a11c877/ : FAILURE in 38m 55s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python3-influxdb/488551f/ : SUCCESS in 43m 28s\n- monasca-tempest-java-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-java-influxdb/18675cd/ : FAILURE in 43m 44s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-cassandra/37108de/ : SUCCESS in 31m 53s\n- build-monasca-common-docker-base-image http://logs.openstack.org/93/659093/1/check/build-monasca-common-docker-base-image/5414072/ : SUCCESS in 2m 11s\n- legacy-monasca-common-maven-build http://logs.openstack.org/93/659093/1/check/legacy-monasca-common-maven-build/00827f3/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4e5fc69f34450536a009f4df5f2612202f8bd00f","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-05-15 08:08:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f96e243ba8f9125d759fee10f0e10adf6fc9c8c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 09:13:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/93/659093/1/check/requirements-check/776b891/ : SUCCESS in 2m 41s\n- openstack-tox-cover http://logs.openstack.org/93/659093/1/check/openstack-tox-cover/8029115/cover/ : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/659093/1/check/openstack-tox-lower-constraints/50eeb37/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/93/659093/1/check/openstack-tox-pep8/c809d20/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/93/659093/1/check/openstack-tox-py27/3a6a5d2/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/93/659093/1/check/openstack-tox-py35/e9d7812/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/93/659093/1/check/openstack-tox-py36/ef9b89d/ : SUCCESS in 3m 57s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-influxdb/23398f4/ : FAILURE in 39m 48s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python3-influxdb/b2d6ac3/ : SUCCESS in 33m 10s\n- monasca-tempest-java-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-java-influxdb/23c4b15/ : FAILURE in 51m 17s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-cassandra/01674e9/ : FAILURE in 41m 19s\n- build-monasca-common-docker-base-image http://logs.openstack.org/93/659093/1/check/build-monasca-common-docker-base-image/d8769d9/ : SUCCESS in 2m 33s\n- legacy-monasca-common-maven-build http://logs.openstack.org/93/659093/1/check/legacy-monasca-common-maven-build/a437f14/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c6c4a7b95e5a09593104b55d8d21d6fb56259fd4","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-05-15 09:19:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c83bbf91bec4a5ed7f4f84c016163877fe70743","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 10:15:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/93/659093/1/check/requirements-check/c61760d/ : SUCCESS in 3m 11s\n- openstack-tox-cover http://logs.openstack.org/93/659093/1/check/openstack-tox-cover/7f63a88/cover/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/659093/1/check/openstack-tox-lower-constraints/254cd51/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/93/659093/1/check/openstack-tox-pep8/eee1058/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/93/659093/1/check/openstack-tox-py27/b979418/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/93/659093/1/check/openstack-tox-py35/1dd5895/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/93/659093/1/check/openstack-tox-py36/410d87d/ : SUCCESS in 3m 26s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-influxdb/cca9fa4/ : SUCCESS in 34m 27s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-python3-influxdb/f1f6a90/ : SUCCESS in 38m 24s\n- monasca-tempest-java-influxdb http://logs.openstack.org/93/659093/1/check/monasca-tempest-java-influxdb/fdf0fae/ : FAILURE in 49m 52s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/93/659093/1/check/monasca-tempest-python2-cassandra/d6164d6/ : SUCCESS in 37m 35s\n- build-monasca-common-docker-base-image http://logs.openstack.org/93/659093/1/check/build-monasca-common-docker-base-image/2726163/ : SUCCESS in 2m 10s\n- legacy-monasca-common-maven-build http://logs.openstack.org/93/659093/1/check/legacy-monasca-common-maven-build/d789715/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"dc381b9bd06dda64a7af34988be8081fda7c5d30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 10:15:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fd4872ee5eaf18a29f0b32e894290200da4b96a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 11:01:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/93/659093/1/gate/requirements-check/14a0582/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/659093/1/gate/openstack-tox-lower-constraints/2820bc3/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/93/659093/1/gate/openstack-tox-pep8/5538e37/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/93/659093/1/gate/openstack-tox-py27/bfb3b56/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/93/659093/1/gate/openstack-tox-py35/9d35ce5/ : SUCCESS in 6m 31s\n- openstack-tox-py36 http://logs.openstack.org/93/659093/1/gate/openstack-tox-py36/9e3e4b9/ : SUCCESS in 4m 28s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/93/659093/1/gate/monasca-tempest-python2-influxdb/3459088/ : SUCCESS in 39m 37s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/93/659093/1/gate/monasca-tempest-python3-influxdb/fdba8e5/ : SUCCESS in 36m 34s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/93/659093/1/gate/monasca-tempest-python2-cassandra/377c7c9/ : SUCCESS in 39m 46s\n- legacy-monasca-common-maven-build http://logs.openstack.org/93/659093/1/gate/legacy-monasca-common-maven-build/a7b48a6/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"12126976738c311490f20cc87e2327380bfada79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 11:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40b808647e46736e438aba5ee6a2006c37c491b1","revisions":{"40b808647e46736e438aba5ee6a2006c37c491b1":{"kind":"REWORK","_number":1,"created":"2019-05-14 15:00:35.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/93/659093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/93/659093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/93/659093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/93/659093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/93/659093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/93/659093/1"}}},"commit":{"parents":[{"commit":"61e2e00454734e2881611abec8df0d85bf7655ac","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/61e2e00454734e2881611abec8df0d85bf7655ac"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-14 14:55:01.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-05-14 14:55:01.000000000","tz":120},"subject":"Blacklist bandit 1.6.0","message":"Blacklist bandit 1.6.0\n\nThere\u0027s a regression[0] in bandit 1.6.0 which causes bandit to stop\nrespecting excluded directories, and our tests throw a bunch of\nviolations. Blacklist this version, but allow newer versions as there is\nalready a pull request[1] to fix it, and it is expected to be included\nin the next release.\n\n[0] https://github.com/PyCQA/bandit/issues/488\n[1] https://github.com/PyCQA/bandit/pull/489\n\nChange-Id: If2e5f82279981ced89da6fc134a238105be60d8b\nStory: 2005676\nTask: 30988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/40b808647e46736e438aba5ee6a2006c37c491b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/40b808647e46736e438aba5ee6a2006c37c491b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
