)]}'
{"id":"openstack%2Fmonasca-common~670472","triplet_id":"openstack%2Fmonasca-common~master~I4b4dc0410715ab9e1f4c7fc786df10ec2f615b59","project":"openstack/monasca-common","branch":"master","topic":"remove-unused-code","hashtags":[],"change_id":"I4b4dc0410715ab9e1f4c7fc786df10ec2f615b59","subject":"Removed unused code","status":"MERGED","created":"2019-07-12 08:21:28.000000000","updated":"2019-07-12 13:40:42.000000000","submitted":"2019-07-12 13:40:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670472-1562938842015-fd40a304","meta_rev_id":"cad52a39dc6a10e9c7ef30de7988782f8044091c","_number":670472,"virtual_id_number":670472,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-12 13:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"date":"2019-07-12 12:13:05.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-12 12:37:30.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":2,"date":"2019-07-12 12:15:14.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 12:37:30.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 12:15:14.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2019-07-12 12:37:30.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-07-12 13:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e95fadc490c264dc09bfc2f78b39464ebc695437","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-12 08:21:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c9ef8a054663213d8b00af5fc082c42dda3f39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 09:16:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/72/670472/1/check/openstack-tox-cover/a1477d6/cover/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/670472/1/check/openstack-tox-lower-constraints/9b19cd6/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/72/670472/1/check/openstack-tox-pep8/2d191bd/ : SUCCESS in 7m 16s\n- openstack-tox-py27 http://logs.openstack.org/72/670472/1/check/openstack-tox-py27/0cf5e40/ : SUCCESS in 6m 59s\n- openstack-tox-py36 http://logs.openstack.org/72/670472/1/check/openstack-tox-py36/5c4b582/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/72/670472/1/check/openstack-tox-py37/ad9a30f/ : SUCCESS in 3m 35s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-influxdb/0b923af/ : SUCCESS in 45m 07s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python3-influxdb/b8a9876/ : SUCCESS in 46m 01s\n- monasca-tempest-java-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-java-influxdb/4918483/ : FAILURE in 52m 00s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-cassandra/7678e02/ : FAILURE in 48m 07s\n- build-monasca-common-docker-base-image http://logs.openstack.org/72/670472/1/check/build-monasca-common-docker-base-image/f1bdf89/ : SUCCESS in 3m 19s\n- legacy-monasca-common-maven-build http://logs.openstack.org/72/670472/1/check/legacy-monasca-common-maven-build/35e0b36/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c04f8900e2333ed416d4cbc088246640e6cca11e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-12 09:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b0ca4a00103be334aa843f9d40b92e8a92330454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 10:24:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/72/670472/1/check/openstack-tox-cover/26097c7/cover/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/670472/1/check/openstack-tox-lower-constraints/4d1f36e/ : SUCCESS in 9m 22s\n- openstack-tox-pep8 http://logs.openstack.org/72/670472/1/check/openstack-tox-pep8/6488cad/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/72/670472/1/check/openstack-tox-py27/e74aa73/ : SUCCESS in 7m 48s\n- openstack-tox-py36 http://logs.openstack.org/72/670472/1/check/openstack-tox-py36/ef30737/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/72/670472/1/check/openstack-tox-py37/0c1623c/ : SUCCESS in 6m 54s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-influxdb/2998f51/ : FAILURE in 39m 06s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python3-influxdb/83d74a3/ : SUCCESS in 43m 13s\n- monasca-tempest-java-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-java-influxdb/3bf6c67/ : FAILURE in 49m 41s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-cassandra/80be402/ : SUCCESS in 39m 43s\n- build-monasca-common-docker-base-image http://logs.openstack.org/72/670472/1/check/build-monasca-common-docker-base-image/97abff2/ : SUCCESS in 2m 54s\n- legacy-monasca-common-maven-build http://logs.openstack.org/72/670472/1/check/legacy-monasca-common-maven-build/13765eb/ : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"110944b2426ce274363e0d7b17824f2fcec94f55","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-12 10:35:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b38c20038e72f3c101648e9324defbfdcf859a2","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-12 11:26:23.000000000","message":"Patch Set 1:\n\nCould we please add missing tests to monasca-api? Please compare my comment in https://review.opendev.org/651208","accounts_in_message":[],"_revision_number":1},{"id":"ecce0256a8109781ff8a89be5a85f48231b801d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 11:30:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/72/670472/1/check/openstack-tox-cover/697f9bf/cover/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/72/670472/1/check/openstack-tox-lower-constraints/7dcf54c/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/72/670472/1/check/openstack-tox-pep8/ec5416a/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/72/670472/1/check/openstack-tox-py27/45a6fa1/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/72/670472/1/check/openstack-tox-py36/798a9d3/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/72/670472/1/check/openstack-tox-py37/c081128/ : SUCCESS in 5m 19s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-influxdb/ca97a88/ : SUCCESS in 38m 26s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-python3-influxdb/efb1fac/ : SUCCESS in 40m 32s\n- monasca-tempest-java-influxdb http://logs.openstack.org/72/670472/1/check/monasca-tempest-java-influxdb/9bd870b/ : FAILURE in 50m 14s (non-voting)\n- monasca-tempest-python2-cassandra http://logs.openstack.org/72/670472/1/check/monasca-tempest-python2-cassandra/5be5037/ : SUCCESS in 41m 19s\n- build-monasca-common-docker-base-image http://logs.openstack.org/72/670472/1/check/build-monasca-common-docker-base-image/5e166f3/ : SUCCESS in 3m 15s\n- legacy-monasca-common-maven-build http://logs.openstack.org/72/670472/1/check/legacy-monasca-common-maven-build/2427abb/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1494dc1b05980232562c11e19f6a327f85ec0f09","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-07-12 12:13:05.000000000","message":"Patch Set 1:\n\n\u003e Could we please add missing tests to monasca-api? Please compare my\n \u003e comment in https://review.opendev.org/651208\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"7449c1f5907b130254b8dd515a85430ab8af1064","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-12 12:15:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"723f5e35f85e6749e2bbec7a7ce11f4e7b035931","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-07-12 12:37:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bd5c3532261cd9ce3a3ec915f8a208f3b7ba845","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 12:37:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65939f293c5428b458c1137245916554ce76dc14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:40:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/72/670472/1/gate/openstack-tox-lower-constraints/755ba96/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/72/670472/1/gate/openstack-tox-pep8/5a7eccd/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/72/670472/1/gate/openstack-tox-py27/9a8f8f3/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/72/670472/1/gate/openstack-tox-py36/5ffdc2c/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/72/670472/1/gate/openstack-tox-py37/baadece/ : SUCCESS in 3m 43s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/72/670472/1/gate/monasca-tempest-python2-influxdb/398ad01/ : SUCCESS in 37m 07s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/72/670472/1/gate/monasca-tempest-python3-influxdb/5f7b474/ : SUCCESS in 37m 42s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/72/670472/1/gate/monasca-tempest-python2-cassandra/8036525/ : SUCCESS in 34m 58s\n- legacy-monasca-common-maven-build http://logs.openstack.org/72/670472/1/gate/legacy-monasca-common-maven-build/eb1ca84/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a4cc9f7634c98abf09d55e3dd26b3d3facce2169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:40:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94bca491ed30aa4bfcf7c4270679ebff27553145","revisions":{"94bca491ed30aa4bfcf7c4270679ebff27553145":{"kind":"REWORK","_number":1,"created":"2019-07-12 08:21:28.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/72/670472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/72/670472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/72/670472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/72/670472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/72/670472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/72/670472/1"}}},"commit":{"parents":[{"commit":"e3ea6d31607e389dd776f897f4607302aa01f84f","subject":"Merge \"In Python 3.7 \"async\" and \"await\" will become reserved keywords\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e3ea6d31607e389dd776f897f4607302aa01f84f"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2019-07-12 08:17:44.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2019-07-12 08:17:44.000000000","tz":120},"subject":"Removed unused code","message":"Removed unused code\n\nThe code of alarm_expr_parser was copied from monasca-api to monasca-common\nin order to be used in monasca-ui, but after more than a year it isn\u0027t used\nin monasca-ui.\n\nChange-Id: I4b4dc0410715ab9e1f4c7fc786df10ec2f615b59\nStory: 2001837\nTask: 35800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/94bca491ed30aa4bfcf7c4270679ebff27553145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/94bca491ed30aa4bfcf7c4270679ebff27553145"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
