)]}'
{"id":"openstack%2Fmonasca-common~671451","triplet_id":"openstack%2Fmonasca-common~master~Iffee218b0a9cac572b6ea13d96da5b8f9e0d0b0d","project":"openstack/monasca-common","branch":"master","topic":"confluent_kafka","hashtags":[],"change_id":"Iffee218b0a9cac572b6ea13d96da5b8f9e0d0b0d","subject":"Fix Confluent Kafka consumer loop","status":"MERGED","created":"2019-07-18 10:22:39.000000000","updated":"2019-08-05 09:44:06.000000000","submitted":"2019-08-05 09:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671451-1564998246418-c33df9d0","meta_rev_id":"21479e343f943d56a0503ab50561118771743ad5","_number":671451,"virtual_id_number":671451,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-05 09:44:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-31 20:48:34.000000000","_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":1,"date":"2019-07-26 17:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"date":"2019-07-31 15:12:37.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},{"value":1,"date":"2019-08-05 08:48:03.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 17:39:45.000000000","updated_by":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"reviewer":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"state":"REVIEWER"},{"updated":"2019-07-31 15:12:37.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-31 20:48:34.000000000","updated_by":{"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},"reviewer":{"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},"state":"REVIEWER"},{"updated":"2019-08-05 08:48:03.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-08-05 09:44:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26b701f73426bb3135ca87af1f24391724d91972","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-18 10:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"267a4a07d350d08ba2c864b1939e689feb4cd7a2","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-18 10:24:36.000000000","message":"Topic set to confluent_kafka","accounts_in_message":[],"_revision_number":1},{"id":"a0c94a3d9aae00f7fcead66913df5fb32b6684b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/51/671451/1/check/openstack-tox-cover/26b5357/cover/ : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/671451/1/check/openstack-tox-lower-constraints/d6b1631/ : SUCCESS in 9m 27s\n- openstack-tox-pep8 http://logs.openstack.org/51/671451/1/check/openstack-tox-pep8/1f056fb/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/51/671451/1/check/openstack-tox-py27/5054e9a/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/51/671451/1/check/openstack-tox-py36/9ecc358/ : SUCCESS in 7m 21s\n- openstack-tox-py37 http://logs.openstack.org/51/671451/1/check/openstack-tox-py37/c187fba/ : SUCCESS in 4m 08s\n- monasca-tempest-python2-influxdb http://logs.openstack.org/51/671451/1/check/monasca-tempest-python2-influxdb/65b72b2/ : SUCCESS in 40m 38s\n- monasca-tempest-python3-influxdb http://logs.openstack.org/51/671451/1/check/monasca-tempest-python3-influxdb/1269f2d/ : SUCCESS in 46m 49s\n- monasca-tempest-python2-cassandra http://logs.openstack.org/51/671451/1/check/monasca-tempest-python2-cassandra/21033f2/ : SUCCESS in 46m 33s\n- build-monasca-common-docker-base-image http://logs.openstack.org/51/671451/1/check/build-monasca-common-docker-base-image/5130f9c/ : SUCCESS in 2m 22s\n- legacy-monasca-common-maven-build http://logs.openstack.org/51/671451/1/check/legacy-monasca-common-maven-build/90ea150/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5401aec387ace60864f67f2c48beb5f64e37f064","author":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"date":"2019-07-26 17:39:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5231dbdfd11d33ee9d91eea7695bec6f16298cb9","author":{"_account_id":26733,"name":"Amir Mofakhar","email":"amir@mofakhar.info","username":"amofakhar"},"date":"2019-07-31 20:48:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66c70a42400092da8a9802ab2b85ed690c66c5c0","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-08-05 08:48:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d80fd7a9bc340cf665e1154430b6ceb915e181a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 08:48:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72db49b4769b8ea9a1021fbc4798534aa1ce762d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:44:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21479e343f943d56a0503ab50561118771743ad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 09:44:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/51/671451/1/gate/openstack-tox-lower-constraints/0e00fc7/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://logs.opendev.org/51/671451/1/gate/openstack-tox-pep8/71e7b4c/ : SUCCESS in 4m 34s\n- openstack-tox-py27 https://logs.opendev.org/51/671451/1/gate/openstack-tox-py27/8ee7d37/ : SUCCESS in 5m 02s\n- openstack-tox-py36 https://logs.opendev.org/51/671451/1/gate/openstack-tox-py36/aba1ecd/ : SUCCESS in 4m 37s\n- openstack-tox-py37 https://logs.opendev.org/51/671451/1/gate/openstack-tox-py37/8cd1a8a/ : SUCCESS in 4m 53s\n- monasca-tempest-python2-influxdb https://logs.opendev.org/51/671451/1/gate/monasca-tempest-python2-influxdb/0be2168/ : SUCCESS in 49m 32s\n- monasca-tempest-python3-influxdb https://logs.opendev.org/51/671451/1/gate/monasca-tempest-python3-influxdb/e173cd6/ : SUCCESS in 46m 00s\n- monasca-tempest-python2-cassandra https://logs.opendev.org/51/671451/1/gate/monasca-tempest-python2-cassandra/00e30eb/ : SUCCESS in 45m 47s\n- legacy-monasca-common-maven-build https://logs.opendev.org/51/671451/1/gate/legacy-monasca-common-maven-build/209a7ca/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce053d1094bcc5035a06ac093f725af49e178dda","revisions":{"ce053d1094bcc5035a06ac093f725af49e178dda":{"kind":"REWORK","_number":1,"created":"2019-07-18 10:22:39.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/51/671451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/51/671451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/51/671451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/51/671451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/51/671451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/51/671451/1"}}},"commit":{"parents":[{"commit":"e4976083a0e6e7ac4c94bb6c075e4db0af405453","subject":"Merge \"Disable monasca-tempest-java-influxdb job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e4976083a0e6e7ac4c94bb6c075e4db0af405453"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-17 21:03:33.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-18 10:16:24.000000000","tz":120},"subject":"Fix Confluent Kafka consumer loop","message":"Fix Confluent Kafka consumer loop\n\nEven when there is no new message consumed the consumer loop should\nproceed to execute the commit callback. Otherwise the batched messages\nwhich which were consumed earlier do not get commited until new message\narrives. Although the case is nearly impossible to happen in real\nenvironment it causes tempest tests to fail when only two messages are\ncreated for testing.\n\nStory: 2003705\nTask: 35855\n\nChange-Id: Iffee218b0a9cac572b6ea13d96da5b8f9e0d0b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ce053d1094bcc5035a06ac093f725af49e178dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/ce053d1094bcc5035a06ac093f725af49e178dda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
