)]}'
{"id":"openstack%2Fmonasca-common~674016","triplet_id":"openstack%2Fmonasca-common~master~I3bf24002aab3d708d045704214a66f5f47863d5d","project":"openstack/monasca-common","branch":"master","topic":"llibrdkafka","hashtags":[],"change_id":"I3bf24002aab3d708d045704214a66f5f47863d5d","subject":"Install librdkafka from source code","status":"MERGED","created":"2019-08-01 12:10:31.000000000","updated":"2019-08-20 07:46:35.000000000","submitted":"2019-08-20 07:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674016-1566287195865-25fc65c6","meta_rev_id":"f80bcdc0fd3b308b10fd99c2e3f90859252c9ba8","_number":674016,"virtual_id_number":674016,"owner":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},{"value":2,"date":"2019-08-20 07:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":2,"date":"2019-08-20 06:57:10.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":1,"date":"2019-08-20 06:57:10.000000000","_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 06:57:10.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-08-20 07:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c470fbdcdef5e21c6a5ffce3c705c0d70e37b33","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-01 12:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6de2109fe4112fd0ffdeff38c6e5a1bafc717fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 12:40:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/16/674016/1/check/openstack-tox-cover/d3dce93/cover/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://logs.opendev.org/16/674016/1/check/openstack-tox-lower-constraints/9b19152/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://logs.opendev.org/16/674016/1/check/openstack-tox-pep8/842827b/ : SUCCESS in 3m 48s\n- openstack-tox-py27 https://logs.opendev.org/16/674016/1/check/openstack-tox-py27/30287ed/ : SUCCESS in 3m 15s\n- openstack-tox-py36 https://logs.opendev.org/16/674016/1/check/openstack-tox-py36/2c87587/ : SUCCESS in 4m 38s\n- openstack-tox-py37 https://logs.opendev.org/16/674016/1/check/openstack-tox-py37/3964424/ : SUCCESS in 3m 56s\n- monasca-tempest-python2-influxdb https://logs.opendev.org/16/674016/1/check/monasca-tempest-python2-influxdb/f55af6c/ : FAILURE in 21m 32s\n- monasca-tempest-python3-influxdb https://logs.opendev.org/16/674016/1/check/monasca-tempest-python3-influxdb/3b8a695/ : FAILURE in 24m 16s\n- monasca-tempest-python2-cassandra https://logs.opendev.org/16/674016/1/check/monasca-tempest-python2-cassandra/42dbfb9/ : FAILURE in 23m 08s\n- build-monasca-common-docker-base-image https://logs.opendev.org/16/674016/1/check/build-monasca-common-docker-base-image/913b44f/ : SUCCESS in 2m 54s\n- legacy-monasca-common-maven-build https://logs.opendev.org/16/674016/1/check/legacy-monasca-common-maven-build/49159b5/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"03fea51fa6f72f773b22a6bdd520565fafd387db","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-14 12:19:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nJuts small comments, other than that I tested it and it\u0027s working fine.\nThank you for this fix.","accounts_in_message":[],"_revision_number":1},{"id":"bb856a0926059b6e307577a6d9a887e7ca8245e6","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-15 00:40:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bfc86cf579030cf46e8f3f95c7dec9c72c98a766","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-15 00:41:24.000000000","message":"Patch Set 1:\n\nok.very thanks","accounts_in_message":[],"_revision_number":1},{"id":"c784fc26dd6cf8fef565b461c174597af65e425b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:26:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099b2443cad34d6aaf2ac77b770eb677 : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91fd686e52764179b1eff135180f5ae9 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e49615e10248b1879651bf9c549fa7 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9381781e1bbf43328002b9c569ad5ac3 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f1e9aa4bd6f4941a85f6559962bc393 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9628d8c5e3834e10a77bfcd414c69c31 : SUCCESS in 5m 25s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/ad226c0d92834869bc7696edd83e85f1 : SUCCESS in 35m 41s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f872eef93ebf4b7a829c5f671ad4bf48 : FAILURE in 36m 33s\n- monasca-tempest-python2-cassandra https://zuul.opendev.org/t/openstack/build/9f38ce2d24ca436ea44ee1e7c84249a0 : SUCCESS in 38m 50s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/c594841d40a249beaa74cfd324a4170f : SUCCESS in 2m 07s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/4c3244a35f22423eab9417ea189a9d33 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3efe04b36a0ebf5b4802396e675def8beac1dfe6","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-19 12:12:43.000000000","message":"Patch Set 2:\n\nLooks like you only rebased this change and not removed this empty lines, they are still there.","accounts_in_message":[],"_revision_number":2},{"id":"f6aabf86afeb40d84297f8636da9b90ecd4164f0","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-20 00:43:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6813974e5840843db17078223bcabf127c8b236","author":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"date":"2019-08-20 00:44:12.000000000","message":"Patch Set 2:\n\nsorry forget to git add,i will review agin.","accounts_in_message":[],"_revision_number":2},{"id":"218907e7f6c9817dd6b49a21ade46b37e7611dfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 01:31:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132bfb2020b648bdb0667fff79e82447 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c79182288c784085a35a5f544f7c2b60 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3132594b2aad442d8730e372f59306b3 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7556e8313e5440980428efe005528c7 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aa48181142f4eca98b39517c98a8f11 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b91a06bd87b44044b31f6c38cc5582b1 : SUCCESS in 4m 12s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/bbfbbc8a722b4afba5a925fcc5a7cd2c : SUCCESS in 37m 58s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4e43836f9e8d4c0da1734dacf28f921e : SUCCESS in 42m 04s\n- monasca-tempest-python2-cassandra https://zuul.opendev.org/t/openstack/build/2fb79142766a4255b5d3ce6601d1f671 : SUCCESS in 37m 58s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/97d49cb5d8644fcdb2bcb0936fc5ca95 : SUCCESS in 2m 39s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/99e41abb59f745a98e5cb29fa9f6292f : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8f21fe75d987c93840de73fe1794a7d2ac716527","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-08-20 06:57:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you.","accounts_in_message":[],"_revision_number":3},{"id":"a9d0691df20ce968bb803917189920cd99078392","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 06:57:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a8cf005a4986a7c7b572169a18a9cf6205a29ca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 07:46:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0bde0b247879f0b747a757f909da07d9c5550ea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 07:46:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50e8636dfaa7476186174fc4588ad4aa : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ac588e6d3947819b05045234c15510 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6892d7bd5504e658c80f39ac033344f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec9d4e2216e44fa798acd36d1b815dce : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18e294f1d0ca4b1fafc7ef36e0778aaa : SUCCESS in 4m 42s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/f7297f262cfb425fbcaf8017f5e2e84f : SUCCESS in 47m 34s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/8f7eff1c1b174848890dd8c7161ec486 : SUCCESS in 35m 05s\n- monasca-tempest-python2-cassandra https://zuul.opendev.org/t/openstack/build/d418231249e14b2f99f1486f18ffb87b : SUCCESS in 40m 22s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/0338a4e6f2174419bff16ec9e84662a9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e9efa3be0e3a67780d7837e0ed2ff7b964a81411","revisions":{"28716223ac4fa0b8e37927952833a2dacf0fb56d":{"kind":"REWORK","_number":1,"created":"2019-08-01 12:10:31.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/16/674016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/16/674016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/1"}}},"commit":{"parents":[{"commit":"e4976083a0e6e7ac4c94bb6c075e4db0af405453","subject":"Merge \"Disable monasca-tempest-java-influxdb job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e4976083a0e6e7ac4c94bb6c075e4db0af405453"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-01 12:07:46.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-01 12:07:46.000000000","tz":480},"subject":"Install librdkafka from source code","message":"Install librdkafka from source code\n\napk install librdkafka-dev success,but loading shared library\nlibrdkafka.so.1 failed.\n\nChange-Id: I3bf24002aab3d708d045704214a66f5f47863d5d\nTask: 36048\nStory: 2006317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/28716223ac4fa0b8e37927952833a2dacf0fb56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/28716223ac4fa0b8e37927952833a2dacf0fb56d"}]},"branch":"refs/heads/master"},"09b9171ab89e891b231e7978bd3c180eb86c10ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-15 00:40:44.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/16/674016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/16/674016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/2"}}},"commit":{"parents":[{"commit":"a6843fdd3b40b1985cc86aec2ec15db60dfca4af","subject":"Merge \"Confluent Kafka driver broker option bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a6843fdd3b40b1985cc86aec2ec15db60dfca4af"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-01 12:07:46.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-15 00:40:39.000000000","tz":480},"subject":"Install librdkafka from source code","message":"Install librdkafka from source code\n\napk install librdkafka-dev success,but loading shared library\nlibrdkafka.so.1 failed.\n\nChange-Id: I3bf24002aab3d708d045704214a66f5f47863d5d\nTask: 36048\nStory: 2006317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/09b9171ab89e891b231e7978bd3c180eb86c10ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/09b9171ab89e891b231e7978bd3c180eb86c10ff"}]},"branch":"refs/heads/master"},"e9efa3be0e3a67780d7837e0ed2ff7b964a81411":{"kind":"REWORK","_number":3,"created":"2019-08-20 00:43:22.000000000","uploader":{"_account_id":29970,"name":"jianweizhang","email":"zhangjianweibj@inspur.com","username":"zhangjianwei"},"ref":"refs/changes/16/674016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/16/674016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/16/674016/3"}}},"commit":{"parents":[{"commit":"a6843fdd3b40b1985cc86aec2ec15db60dfca4af","subject":"Merge \"Confluent Kafka driver broker option bug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a6843fdd3b40b1985cc86aec2ec15db60dfca4af"}]}],"author":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-01 12:07:46.000000000","tz":480},"committer":{"name":"zhangjianweibj","email":"zhangjianweibj@inspur.com","date":"2019-08-20 00:42:51.000000000","tz":480},"subject":"Install librdkafka from source code","message":"Install librdkafka from source code\n\napk install librdkafka-dev success,but loading shared library\nlibrdkafka.so.1 failed.\n\nChange-Id: I3bf24002aab3d708d045704214a66f5f47863d5d\nTask: 36048\nStory: 2006317\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e9efa3be0e3a67780d7837e0ed2ff7b964a81411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/e9efa3be0e3a67780d7837e0ed2ff7b964a81411"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
