)]}'
{"id":"openstack%2Fmonasca-common~694738","triplet_id":"openstack%2Fmonasca-common~master~I7627a8caa2d6bb9c7789df143e72a4085060b164","project":"openstack/monasca-common","branch":"master","topic":"story_2006059","hashtags":[],"change_id":"I7627a8caa2d6bb9c7789df143e72a4085060b164","subject":"Allow passing config options to Kafka producer","status":"MERGED","created":"2019-11-18 10:13:02.000000000","updated":"2019-12-10 10:48:32.000000000","submitted":"2019-12-10 10:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694738-1575974912372-349ca902","meta_rev_id":"ccadc1b5319f186aeefe8f5c63552979fc794b2a","_number":694738,"virtual_id_number":694738,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 10:48:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"date":"2019-11-18 12:11:19.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"all":[{"value":1,"date":"2019-11-25 13:14:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":2,"date":"2019-11-18 10:40:41.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-10 10:10:02.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"all":[{"value":0,"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-10 10:10:02.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 10:40:41.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 13:14:55.000000000","updated_by":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"reviewer":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"state":"REVIEWER"},{"updated":"2019-12-10 10:10:02.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2019-12-10 10:48:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"017fd627a7e711b5b732f8ebb0736e8dccbca966","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-18 10:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa98fcf44c8c2d76cf768dd3c2e5e109a7eec812","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-18 10:40:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74716c75ed580cf0c92c75486d9d05e57fc4806f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:00:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7f50504629a4802b72a4f9e15108560 : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad350005b0c453ca1f079078e6dfd5f : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f86b2abd4345379b1ea739cf68a657 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2563962ce130434c87dff43a742d5abb : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f047cdcab5b4b2f968264f2ebac8782 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/817fe569abd0479198ec05a9e53fcea6 : SUCCESS in 6m 03s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/604c77bda3dd470d9732a76984051642 : SUCCESS in 44m 47s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/2e1bb12dfa8f45029330bd02aa7569f8 : SUCCESS in 39m 46s\n- monasca-tempest-python2-cassandra https://zuul.opendev.org/t/openstack/build/abcb47fc43864608bef78ef08b7a10d5 : FAILURE in 42m 27s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/1b0d6031d98047668d42ff2c4450661d : SUCCESS in 2m 16s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/505eb6c79f304c518bf19600d490c368 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b1fb3d1fd16baa2e4b7bf58099bac2b7f54aee2a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-18 12:11:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf777d7791b9901609452a9bb1e846736753f92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:02:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b716ab279714f4dbcd86dbff97fc331 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6c204a6dbdb40be88a01e37d8b18148 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdbf93e043d498fbd3ae5d3110d5a9f : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69ef74fe2d554a95954ece695016389d : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f977e3843c884b329d7dcb83517fb8b7 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28f7f72564794b469958acec3797f69d : SUCCESS in 4m 03s\n- monasca-tempest-python2-influxdb https://zuul.opendev.org/t/openstack/build/b983d619cdc74a589d9094be35975bf8 : SUCCESS in 39m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f11c92c927804f3baec14d3c59374e12 : SUCCESS in 48m 05s\n- monasca-tempest-python2-cassandra https://zuul.opendev.org/t/openstack/build/7f97189d689f412e8de9c39ecc7b2851 : SUCCESS in 48m 21s\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/72403bd818a04fba86e800b14afb16c3 : SUCCESS in 3m 03s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/4ab382c0794d4d7c88890ce957690758 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"574792419a5f11c83a7090ec74857a976beba669","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-20 12:13:35.000000000","message":"Topic set to story_2006059","accounts_in_message":[],"_revision_number":1},{"id":"5440f6e478b824ba53a61c37ca9e9e9784a5046e","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2019-11-25 13:14:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1a2797e0bcae74ddb1e1f40647a0bc45f600c21","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2019-12-10 10:10:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5debe5d43c5c906888e8734b0fcbb5de5d979292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:10:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42528138ecffce50adc2c12bebb6737aed91582c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:48:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cdcaa8d5c324b48863f36f24b6fc17c : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13da33a5b58e4fbabbbe1d963937952a : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb017e0336c84682bdfe9cd93243a856 : SUCCESS in 6m 07s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/fd3a9c37c7254cb38a5f64a11b2fb6a0 : SUCCESS in 36m 26s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/c60ffd13b3dd49af913ec275850bb086 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"43f8abfbd0ffa382feae0600ea00b2bae6272610","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 10:48:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb8e6ff7e259e68ec35ce0e6e1345dd6e7cfb7a4","revisions":{"eb8e6ff7e259e68ec35ce0e6e1345dd6e7cfb7a4":{"kind":"REWORK","_number":1,"created":"2019-11-18 10:13:02.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/38/694738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/38/694738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/38/694738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/38/694738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/38/694738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/38/694738/1"}}},"commit":{"parents":[{"commit":"66754b1a4aa2efee615eebae4a9da191bc6764a5","subject":"Avoid value decoding in Kafka delivery report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/66754b1a4aa2efee615eebae4a9da191bc6764a5"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-18 10:06:12.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-18 10:06:12.000000000","tz":60},"subject":"Allow passing config options to Kafka producer","message":"Allow passing config options to Kafka producer\n\nThe change allows passing supported options as dictionary to create a\nKafka producer object.\n\nChange-Id: I7627a8caa2d6bb9c7789df143e72a4085060b164\nStory: 2006059\nTask: 37531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/eb8e6ff7e259e68ec35ce0e6e1345dd6e7cfb7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/eb8e6ff7e259e68ec35ce0e6e1345dd6e7cfb7a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
