)]}'
{"id":"openstack%2Fmonasca-common~696042","triplet_id":"openstack%2Fmonasca-common~master~I7fcf30d848efb8dc4cc6ca19056868aa6127cf60","project":"openstack/monasca-common","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I7fcf30d848efb8dc4cc6ca19056868aa6127cf60","subject":"Enable running unit tests in py37 environment","status":"MERGED","created":"2019-11-26 09:19:31.000000000","updated":"2019-12-10 20:18:57.000000000","submitted":"2019-12-10 20:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696042-1576009137838-656d7ee0","meta_rev_id":"506a325a0b266b284a841e90e45a18b4e066f1b5","_number":696042,"virtual_id_number":696042,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 20:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-11-28 09:53:00.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-05 20:55:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-12-10 17:10:56.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 20:55:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-12-10 17:10:56.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-12-10 20:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b700fc05eec25623ca791f8a118130738e9337e3","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-26 09:19:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"875953324402f80d175411bf8108599d95f788c2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-26 09:20:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"62f60ee15fdd997be314c2fc075cd355dee55feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 10:09:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb795af56d144bf282c106b205d3756f : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9078c44f24fd4f2981f0c0fb80375f12 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7a8ba1322744371830eb8432127eaca : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1096758f80014fae8f3969c7eebfa96a : FAILURE in 4m 50s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3e58e34cd99d410781d06a29788412a2 : SUCCESS in 43m 58s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/b159f852ea7c4ca68f32a94ddb0f64a2 : FAILURE in 21m 51s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/2aeef05cdfd848c9899cb3d6fd534351 : SUCCESS in 2m 24s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/a393366155fb46c1aad7cde6c4787d54 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a9239db84341e3d70613a5b329d9f65a330376c2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-26 10:40:08.000000000","message":"Patch Set 1: -Code-Review\n\nLooks like we\u0027ve caught some py37 test failures","accounts_in_message":[],"_revision_number":1},{"id":"30369859d10cb5f1d43976cf4f62ff67c5538b12","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-26 11:39:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e4d6514c1e863ac7ed008f5b3055e92d5816081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:33:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63d87b89d298407cba8e6c7bf8cca910 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1645a1db987945b784f683d4f4b424e0 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ccbf12bd92b4c70a206998a3ee89b6d : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1e50802e11e4b90b3d69de636c5d88a : FAILURE in 4m 18s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f7af9dbc0c8a43678b5795d1e225787a : SUCCESS in 44m 38s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/fa93db161ac24f3e9a28420934d2a9c6 : FAILURE in 28m 14s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/603a940409b245cab0d903a167cbf809 : SUCCESS in 2m 27s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/a0112d0b851b417d81093aca6c726000 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"97f973653afd5bd507af5e6e81a59ee00f95fdef","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-26 15:55:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7da326a40bdd009d37861ad2e34923b7faae2392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:39:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7f592c402ae4bfbb4274d61406c685a : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5409571d47c41cbaca9fe465ad0a76d : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7631c8c7c12a43ff9748da605ec45ab1 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1e640e78ad9459f97d8cb8588b7e03f : FAILURE in 5m 48s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/1afca711d652411e86d8da41387eaf0c : SUCCESS in 40m 56s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/d9e1841d074a40708685d8d064243cbc : FAILURE in 9m 26s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/0eb21d4697744133b4c0d54836be1bfe : SUCCESS in 1m 52s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/0c023ef587ca4869bb5244a8f131b05a : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"36ca16c35d8bd493a8203ca405dd51aa9a56f97c","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-11-27 08:19:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e745c59323bf7e608bae257b3ab75450a185e22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 09:02:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a72794fa4876407ca0432ebc175f791b : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43e4480d092a470097caf340f0dc42d5 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da3a0b5deb994a6d8043628bf45f6d74 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4e44e0b8c294e628cc310d5b0655b08 : SUCCESS in 3m 51s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0a177a293e7b4f37861bce562e1eb735 : SUCCESS in 37m 18s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/69d14dff8f0245be9e4eee461e4b6198 : FAILURE in 25m 30s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/952b3689d0f44e8483d60086debc1e57 : SUCCESS in 2m 06s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/9bf547fe01a94a08af0c12a78d7df841 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"804c3fbe4a0329afcfbe569a7df390a953cae53a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-28 09:53:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"700b66f344a07967a76e139872dbf35a8681c858","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-05 20:55:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0c8241f8668cfaa883137a741b3eb703eadd370a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-10 17:10:56.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f8fa57ea00ddf46dc7fda921bb03889b95c2b45f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:11:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"26dc577afc96f61a8e2db47d143197c1e8f6f271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 18:42:25.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70c98481ac396b368e7a8976f18607454f62cb9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 19:24:55.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"df46c41bccd75aa77a0e31fd1cf34b3faa6b5c09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 20:18:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0620f5f66440479c97c7671a22c3cee4 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0264059ba564715a77a229f159ea807 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c82fd409126f41139442048b45f295a1 : SUCCESS in 5m 58s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/ad666787b6d84c5984420cd76e5bdab4 : SUCCESS in 42m 48s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/54608dc0e37d450ea622b09011feb086 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":4},{"id":"061ba8d5ac32de7d8343dc4e821cda9582d14e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 20:18:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"231f3b4760fe11e1094fa8d5bd24dd2d56475d3d","revisions":{"3c1db9ca392a2ab937c788b34546abf77db5ea94":{"kind":"REWORK","_number":1,"created":"2019-11-26 09:19:31.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/42/696042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/42/696042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/1"}}},"commit":{"parents":[{"commit":"b70afb9bf6538748f74262431e3342cc77b9369e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b70afb9bf6538748f74262431e3342cc77b9369e"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 09:18:59.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 09:18:59.000000000","tz":60},"subject":"Enable running unit tests in py37 environment","message":"Enable running unit tests in py37 environment\n\nUntil now no unit tests have been executed in py37 environment. This\nchange fixes tox configuration to enable running these tests.\n\nChange-Id: I7fcf30d848efb8dc4cc6ca19056868aa6127cf60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3c1db9ca392a2ab937c788b34546abf77db5ea94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3c1db9ca392a2ab937c788b34546abf77db5ea94"}]},"branch":"refs/heads/master"},"960de17fe5051711b248429436ff6f1d08323b17":{"kind":"REWORK","_number":2,"created":"2019-11-26 11:39:50.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/42/696042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/42/696042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/2"}}},"commit":{"parents":[{"commit":"b70afb9bf6538748f74262431e3342cc77b9369e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b70afb9bf6538748f74262431e3342cc77b9369e"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 09:18:59.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 11:36:21.000000000","tz":60},"subject":"Enable running unit tests in py37 environment","message":"Enable running unit tests in py37 environment\n\nUntil now no unit tests have been executed in py37 environment. This\nchange fixes tox configuration to enable running these tests.\n\nIt also fixes two unit tests to handle PEP479 being enabled for all code\nin Python 3.7 [1, 2].\n\n[1] https://docs.python.org/3/whatsnew/3.7.html#changes-in-python-behavior\n[2] https://stackoverflow.com/questions/51700960/runtimeerror-generator-raised-stopiteration-every-time-i-try-to-run-app\n\nChange-Id: I7fcf30d848efb8dc4cc6ca19056868aa6127cf60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/960de17fe5051711b248429436ff6f1d08323b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/960de17fe5051711b248429436ff6f1d08323b17"}]},"branch":"refs/heads/master"},"70654a74ca8ffefed98eccb057c7c671e0a8bf08":{"kind":"REWORK","_number":3,"created":"2019-11-26 15:55:57.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/42/696042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/42/696042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/3"}}},"commit":{"parents":[{"commit":"b70afb9bf6538748f74262431e3342cc77b9369e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b70afb9bf6538748f74262431e3342cc77b9369e"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 09:18:59.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 15:55:45.000000000","tz":60},"subject":"Enable running unit tests in py37 environment","message":"Enable running unit tests in py37 environment\n\nUntil now no unit tests have been executed in py37 environment. This\nchange fixes tox configuration to enable running these tests.\n\nIt also fixes two unit tests to handle PEP479 being enabled for all code\nin Python 3.7 [1, 2].\n\n[1] https://docs.python.org/3/whatsnew/3.7.html#changes-in-python-behavior\n[2] https://stackoverflow.com/questions/51700960/runtimeerror-generator-raised-stopiteration-every-time-i-try-to-run-app\n\nChange-Id: I7fcf30d848efb8dc4cc6ca19056868aa6127cf60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/70654a74ca8ffefed98eccb057c7c671e0a8bf08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/70654a74ca8ffefed98eccb057c7c671e0a8bf08"}]},"branch":"refs/heads/master"},"231f3b4760fe11e1094fa8d5bd24dd2d56475d3d":{"kind":"REWORK","_number":4,"created":"2019-11-27 08:19:53.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/42/696042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/42/696042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/42/696042/4"}}},"commit":{"parents":[{"commit":"b70afb9bf6538748f74262431e3342cc77b9369e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/b70afb9bf6538748f74262431e3342cc77b9369e"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-26 09:18:59.000000000","tz":60},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-11-27 08:19:36.000000000","tz":60},"subject":"Enable running unit tests in py37 environment","message":"Enable running unit tests in py37 environment\n\nUntil now no unit tests have been executed in py37 environment. This\nchange fixes tox configuration to enable running these tests.\n\nIt also fixes two unit tests to handle PEP479 being enabled for all code\nin Python 3.7 [1, 2].\n\n[1] https://docs.python.org/3/whatsnew/3.7.html#changes-in-python-behavior\n[2] https://stackoverflow.com/questions/51700960/runtimeerror-generator-raised-stopiteration-every-time-i-try-to-run-app\n\nChange-Id: I7fcf30d848efb8dc4cc6ca19056868aa6127cf60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/231f3b4760fe11e1094fa8d5bd24dd2d56475d3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/231f3b4760fe11e1094fa8d5bd24dd2d56475d3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
