)]}'
{"id":"openstack%2Fmonasca-common~716925","triplet_id":"openstack%2Fmonasca-common~master~Ic18f8cf4addfedd9055999ec685e81140ffda32d","project":"openstack/monasca-common","branch":"master","topic":"ujson-replacement","hashtags":[],"change_id":"Ic18f8cf4addfedd9055999ec685e81140ffda32d","subject":"Replacing ujson with simplejson","status":"MERGED","created":"2020-04-02 10:43:35.000000000","updated":"2020-04-03 08:58:29.000000000","submitted":"2020-04-03 08:58:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716925-1585904309434-73754ddb","meta_rev_id":"aee738b042be217d84a16e339a0ad017a6cbc992","_number":716925,"virtual_id_number":716925,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 08:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"date":"2020-04-02 14:52:50.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-04-03 08:07:46.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 15:33:34.000000000","_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-03 08:07:46.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 13:01:06.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-04-02 15:33:34.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-04-03 08:07:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-03 08:58:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07bf4f2eec4bfcec0772310a344188734e418e13","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-04-02 10:43:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3b48ec86ca90a3b5c04ff83744cea4f774e6f1","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-04-02 11:08:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab0555abc6c5dd08599337b27f0cff53e8091ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:25:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/092bcd1a644942c5ba35d01bd61222a6 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6bc0dc3443d45cbb6de661770f92f0b : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9976a4e66478455bbaebf95a1ac77b5d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab10261c88874936be8c49ace6e092f3 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b098efbace34f6881a095d09ab2994a : SUCCESS in 6m 13s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/09cceb558b1d4524b03d5c636cc76a8e : SUCCESS in 35m 25s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/845a4ff169a34cf8841b93e64aceaf49 : FAILURE in 32m 27s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/d4331554aa65448aae65fc640e8312a7 : SUCCESS in 3m 51s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/cdc2758ecb7643a49398ed8983afee19 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d4ef21fb3b9c7dd0476e90eadb6e806fccdf085d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-04-02 13:01:06.000000000","message":"Patch Set 1: Code-Review-1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"455f10be0d2aac30a9b65675359d67cfa712308a","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-04-02 13:44:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5b0f7cb8f24c42b4afc42ee99a0e4e87c37d2a73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:49:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b4a7691b0014a39ae914fc801694722 : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89edbbb4f94d4131b6f2a154fc93b6e3 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99d4956fb2e7411aabf5a36a02c170ce : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f4de62820b24695a77c97f573135612 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ae1fb6c9f74deda3a7acf1c1bd6f22 : SUCCESS in 7m 08s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/f1e73b974efb4fcd9fae8bae10f036fc : SUCCESS in 44m 23s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/14d894753d8144d79aa662e529f301bd : FAILURE in 36m 33s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/c4f26b1087e84ef1941a8434326a8fa2 : SUCCESS in 4m 09s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/32e829af0d784f11aa442fedf53b138e : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"08e5052079a2d30c5eed7ffcf7cd73c800e89d32","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-04-02 14:52:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69a3caf0c2841d0a046ffbb254117126d6ff1429","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-04-02 15:33:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a60538e454fd345f20f65bcf1e2a9bcc9f74317","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-03 08:07:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Martin!","accounts_in_message":[],"_revision_number":2},{"id":"f9bde0521336fc10db627e6e467e45bfb37cfc20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:07:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7f7c7be1477850eff74a56c4310c58c7c399c61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:58:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3ef3873e5ca18bc07f513bacdacc77473e7803f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:58:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/270565d3ce52466dbff365c01cd00f3d : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aae0fe9670164f97b56b66cbc014da68 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf66687fd5a42d3871b23fce189ab58 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b07dceac0e5401086d32b7b724d5a94 : SUCCESS in 8m 49s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/4c226ff001734d68ae1c2625bd3e531b : SUCCESS in 43m 54s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/7728fd77f345497c8d62ff8913220034 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f4e2bc5dfbd06f39961123750b3f2f65a5337da","revisions":{"aa8866eb75ffec5b6a7b7f34552bd89ffeba12ec":{"kind":"REWORK","_number":1,"created":"2020-04-02 10:43:35.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/25/716925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/25/716925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/1"}}},"commit":{"parents":[{"commit":"0447757f4e80d4623656bb17e49b707b7618cb54","subject":"fix a typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/0447757f4e80d4623656bb17e49b707b7618cb54"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-04-02 10:39:21.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-04-02 10:39:21.000000000","tz":120},"subject":"Replacing ujson with simplejson","message":"Replacing ujson with simplejson\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-January/012285.html\n\nChange-Id: Ic18f8cf4addfedd9055999ec685e81140ffda32d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/aa8866eb75ffec5b6a7b7f34552bd89ffeba12ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/aa8866eb75ffec5b6a7b7f34552bd89ffeba12ec"}]},"branch":"refs/heads/master"},"0f4e2bc5dfbd06f39961123750b3f2f65a5337da":{"kind":"REWORK","_number":2,"created":"2020-04-02 13:44:53.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/25/716925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/25/716925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/25/716925/2"}}},"commit":{"parents":[{"commit":"0447757f4e80d4623656bb17e49b707b7618cb54","subject":"fix a typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/0447757f4e80d4623656bb17e49b707b7618cb54"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-04-02 10:39:21.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-04-02 13:44:25.000000000","tz":120},"subject":"Replacing ujson with simplejson","message":"Replacing ujson with simplejson\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-January/012285.html\n\nChange-Id: Ic18f8cf4addfedd9055999ec685e81140ffda32d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/0f4e2bc5dfbd06f39961123750b3f2f65a5337da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/0f4e2bc5dfbd06f39961123750b3f2f65a5337da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
