)]}'
{"id":"openstack%2Fmonasca-common~725258","triplet_id":"openstack%2Fmonasca-common~master~I005c16dc2c2c11548d4cdce023599fb9be03a52f","project":"openstack/monasca-common","branch":"master","hashtags":[],"change_id":"I005c16dc2c2c11548d4cdce023599fb9be03a52f","subject":"Bump librdkafka to 1.4.0","status":"MERGED","created":"2020-05-04 12:47:39.000000000","updated":"2020-05-07 13:54:51.000000000","submitted":"2020-05-07 13:54:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725258-1588859691527-92884968","meta_rev_id":"a09afe185dfd4d87f47e347e0cad8f4d96505db2","_number":725258,"virtual_id_number":725258,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 13:54:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":2,"date":"2020-05-07 13:03:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},{"value":1,"date":"2020-05-07 13:03:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 12:59:26.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-05-07 13:03:48.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-07 13:54:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47758bd2bd76e70296360db48f4e541efaf20a37","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-04 12:47:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8cf4f28eb7b93e21838f64053dc6f0e24199c6f","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-04 12:49:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"05ed05524cddd23494e53aa6c025df908c193ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 13:34:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/065b03093ac049d985ae33a190e71477 : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79e8b6c946434e8fb69001d791efb13d : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9dba88d12954d76a96acfb45c8ed1d5 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9080fc4c110b41f48a384a023e7d12ae : SUCCESS in 6m 02s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/1e6cbdaf3d214018978d2133cbd6b3e2 : FAILURE in 39m 33s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/9887ca9eb88c47fd99b58e895fd8a37d : FAILURE in 37m 03s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/f4d791712a8448ecb3add20d1919a9bb : SUCCESS in 2m 42s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/8c26fa71d6d24841be5c1ebcf4ee50a0 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"46f2fe7bd39ce39df0aff5b7ece00acc7239a795","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-04 13:41:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"be26a74f1904a1d8b7e3dbe47eb6999b1b541a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:28:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3053a3bd1734749a7e502781d0c9c4d : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0af797fde3804dd2a76f3010ca826237 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebba28c587b34745b40c47b588b4ac1b : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ac6e87dcc3b43ef9819c115d5bcf709 : SUCCESS in 8m 54s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/01ffd8eb7c324ff79d291a23eacff57c : FAILURE in 40m 43s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/be776eb28c0d4a10a5255f6c108a580f : FAILURE in 33m 38s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/2d203ec6c27348b8a460c9b9b40d0866 : SUCCESS in 3m 22s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/663689cf52334d8e944c0327f64c4261 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"dc563b537f97e005f2ca5d8c6cc38e4e4e0f5c91","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-04 16:43:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"581dba17b223bf22322282d35ce12d33c524d5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 17:33:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80eed4e8605e4193a7534f88f2c8e388 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72680acdd2f04c49882ff890ded43fec : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91c334cefd6749158e35f88a5ddab5ab : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a1abee1f20945ef830492523b3f64b6 : SUCCESS in 6m 40s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/efb621ed57d54508b516dfe212d5ec56 : FAILURE in 38m 20s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/721398fe38e44db6ac04a3adf029b3fd : FAILURE in 40m 26s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/3645bc87192f402bb3444da6765421a6 : SUCCESS in 2m 39s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/003119dea9ef461e949445f2aa29fa18 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3937f3b44ccbe564eed4b6ad8e04c7742f2dfa68","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-07 11:59:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"13056717db6415b146ea2dc3aee81c199f3cf178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 12:48:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26ea0a621504304bc6ef15d2e4943fc : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6308b0621880432c930aa46b67c72d34 : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/733139c1448b4297a4ba32c59fd03b0f : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec33bcfc81be4a0cb9133e68f08efc71 : SUCCESS in 12m 07s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/c15209db5277448ca8cea8749850e8fa : SUCCESS in 47m 24s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/88bf175994e4404db450521d17141b8a : FAILURE in 40m 57s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/31945443cdc94846a53cd71d1cc17e7b : SUCCESS in 2m 20s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/415ff925be1c4f4ebd507b160b70ed89 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":3},{"id":"3939996ca7760c007ffedffc0e27d50e1e6f4882","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-07 13:03:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"444e92a5b6e6acd07f011ea96e09d3755520751d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:03:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1121f199521800a84e26b85bea51b3e1ec0fd26e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:54:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a09afe185dfd4d87f47e347e0cad8f4d96505db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:54:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d62c3e3b79524b48902026e6d9c93a61 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b448681019ea42c089b169a7c63a3323 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fb112912b15418684505fa3408f0b66 : SUCCESS in 5m 44s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/26e15bfe926841a3a5214386687cab06 : SUCCESS in 43m 58s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/f488b6a9d878499cbc5ea3f062a34fc3 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1c20a58b22920c4df7385b240601fe169189fe01","revisions":{"bb8961d3e6e17a77d0f8340ce108eb60a2cacb2b":{"kind":"REWORK","_number":1,"created":"2020-05-04 12:47:39.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/58/725258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/58/725258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/1"}}},"commit":{"parents":[{"commit":"2ebc766534eba6163e98b94a1f114ece18739fff","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2ebc766534eba6163e98b94a1f114ece18739fff"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 11:25:44.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 11:25:44.000000000","tz":120},"subject":"Bump librdkafka to 1.4.1","message":"Bump librdkafka to 1.4.1\n\nPackage confluent-kafka-python requires librdkafka \u003e\u003d 1.4.0 [1]\n\n[1] https://github.com/confluentinc/confluent-kafka-python/tree/v1.4.1#prerequisites\n\nChange-Id: I005c16dc2c2c11548d4cdce023599fb9be03a52f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bb8961d3e6e17a77d0f8340ce108eb60a2cacb2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/bb8961d3e6e17a77d0f8340ce108eb60a2cacb2b"}]},"branch":"refs/heads/master"},"a4eb44b2d966daff8df63b7be70abdad7f114f72":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-04 12:49:21.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/58/725258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/58/725258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/2"}}},"commit":{"parents":[{"commit":"2ebc766534eba6163e98b94a1f114ece18739fff","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2ebc766534eba6163e98b94a1f114ece18739fff"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 11:25:44.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 12:48:41.000000000","tz":120},"subject":"Bump librdkafka to 1.4.0","message":"Bump librdkafka to 1.4.0\n\nPackage confluent-kafka-python v1.4.1 requires librdkafka \u003e\u003d 1.4.0 [1]\n\n[1] https://github.com/confluentinc/confluent-kafka-python/tree/v1.4.1#prerequisites\n\nChange-Id: I005c16dc2c2c11548d4cdce023599fb9be03a52f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a4eb44b2d966daff8df63b7be70abdad7f114f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a4eb44b2d966daff8df63b7be70abdad7f114f72"}]},"branch":"refs/heads/master"},"1c20a58b22920c4df7385b240601fe169189fe01":{"kind":"REWORK","_number":3,"created":"2020-05-04 16:43:19.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/58/725258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/58/725258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/58/725258/3"}}},"commit":{"parents":[{"commit":"2ebc766534eba6163e98b94a1f114ece18739fff","subject":"Update hacking for Python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/2ebc766534eba6163e98b94a1f114ece18739fff"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 11:25:44.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-04 16:42:14.000000000","tz":120},"subject":"Bump librdkafka to 1.4.0","message":"Bump librdkafka to 1.4.0\n\nPackage confluent-kafka-python v1.4.1 requires librdkafka \u003e\u003d 1.4.0 [1]\nBump sublibraries libressl3.0-libcrypto libressl3.0-libssl\n\n[1] https://github.com/confluentinc/confluent-kafka-python/tree/v1.4.1#prerequisites\n\nChange-Id: I005c16dc2c2c11548d4cdce023599fb9be03a52f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1c20a58b22920c4df7385b240601fe169189fe01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/1c20a58b22920c4df7385b240601fe169189fe01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
