)]}'
{"id":"openstack%2Fmonasca-common~729141","triplet_id":"openstack%2Fmonasca-common~master~Id7b8f11058e113103113603ec1f3bdabef4ca6ae","project":"openstack/monasca-common","branch":"master","hashtags":[],"change_id":"Id7b8f11058e113103113603ec1f3bdabef4ca6ae","subject":"Update the outdated link","status":"MERGED","created":"2020-05-19 07:08:40.000000000","updated":"2020-05-19 10:07:36.000000000","submitted":"2020-05-19 10:07:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729141-1589882856973-d1623aff","meta_rev_id":"248bf0757cf6a43a54c92ed160ab1d448ceea1fa","_number":729141,"virtual_id_number":729141,"owner":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 10:07:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-19 09:57:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-19 09:57:48.000000000","_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 09:57:48.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-19 10:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"626654adc335ad12f0d8dadd9b74849870a0bc15","author":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"date":"2020-05-19 07:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"705ec8ff111d205dbc2cfefde0412c394a047c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 07:16:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee2a45f74461415ab18316367510deb5 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4876e5d1e8174e7481d42cbfe0f9295b : SUCCESS in 6m 44s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/60813ad92bd34251b1a1f0bfaf1ce2e6 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1b0846cee6083e639a4792fd9ee29403e2b84be3","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-19 09:57:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":1},{"id":"8b24f892c1264ecd4f3d4a67017552a7facff6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:58:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"faa6a10d706c14ba289a6dfbecc10275455be90f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 10:07:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"248bf0757cf6a43a54c92ed160ab1d448ceea1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 10:07:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0cd801eb1ef4b0dbee94cb7401299db : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2df69ec76b84894b0d6e1344c81ffb0 : SUCCESS in 4m 10s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/a93af2db52cf4521859ce2255a5a7d93 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"712d27b39499146a135a4983cc0c6d014bb6566e","revisions":{"712d27b39499146a135a4983cc0c6d014bb6566e":{"kind":"REWORK","_number":1,"created":"2020-05-19 07:08:40.000000000","uploader":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"ref":"refs/changes/41/729141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/41/729141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/41/729141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/41/729141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/41/729141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/41/729141/1"}}},"commit":{"parents":[{"commit":"850710c46f01957f569f2b86c7151f8c4af7ac6a","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/850710c46f01957f569f2b86c7151f8c4af7ac6a"}]}],"author":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 07:07:03.000000000","tz":480},"committer":{"name":"lidong","email":"lidongbj@inspur.com","date":"2020-05-19 07:08:34.000000000","tz":480},"subject":"Update the outdated link","message":"Update the outdated link\n\nChange-Id: Id7b8f11058e113103113603ec1f3bdabef4ca6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/712d27b39499146a135a4983cc0c6d014bb6566e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/712d27b39499146a135a4983cc0c6d014bb6566e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
