)]}'
{"id":"openstack%2Fmonasca-common~729276","triplet_id":"openstack%2Fmonasca-common~master~I9aafe78878630c0437ca7cac418d87c7e27b1150","project":"openstack/monasca-common","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I9aafe78878630c0437ca7cac418d87c7e27b1150","subject":"Add CONTRIBUTING.rst","status":"MERGED","created":"2020-05-19 14:17:13.000000000","updated":"2020-05-19 20:11:44.000000000","submitted":"2020-05-19 20:11:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729276-1589919104098-cbaaa819","meta_rev_id":"4d7311c1aa0769c2dc144705c36b5807c874d5cb","_number":729276,"virtual_id_number":729276,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 20:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-05-19 20:01:11.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-05-19 20:01:11.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 20:01:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-19 20:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca4bb9bf5701a3dadb9e697087e2b8d2b4ecc878","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-19 14:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e18ef5eb16c10b1b2ac0fbd3aeeb15b6a31715b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 14:29:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62b3401861343f2a34d71c493b55864 : SUCCESS in 7m 15s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/e1352b0613bd409681a672ab4058836e : SUCCESS in 8m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6cffe4e9f93ed41fb01c0cefc07ee4b8fffe7458","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-19 14:52:01.000000000","message":"Topic set to project-ptl-and-contrib-docs","accounts_in_message":[],"_revision_number":1},{"id":"af2413be6ef9791f37f44e42422a9ddbc4408315","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-19 20:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"173b0838dad771907001aea9f55320672b08acfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:01:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0bcf667c85545e3bc05b842740da02b6e0576a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:11:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99be6397e234ed5b0bd61ccfae06aba : SUCCESS in 7m 32s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/6f78a3ffd37349ed83aedfb583c5ccb6 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4d7311c1aa0769c2dc144705c36b5807c874d5cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 20:11:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de460c8946d30be19bd4f7ecb213c2b8894d1a5c","revisions":{"de460c8946d30be19bd4f7ecb213c2b8894d1a5c":{"kind":"REWORK","_number":1,"created":"2020-05-19 14:17:13.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/76/729276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/76/729276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/729276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/729276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/729276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/76/729276/1"}}},"commit":{"parents":[{"commit":"712d27b39499146a135a4983cc0c6d014bb6566e","subject":"Update the outdated link","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/712d27b39499146a135a4983cc0c6d014bb6566e"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-19 14:16:52.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-05-19 14:16:52.000000000","tz":120},"subject":"Add CONTRIBUTING.rst","message":"Add CONTRIBUTING.rst\n\nChange-Id: I9aafe78878630c0437ca7cac418d87c7e27b1150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/de460c8946d30be19bd4f7ecb213c2b8894d1a5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/de460c8946d30be19bd4f7ecb213c2b8894d1a5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
