)]}'
{"id":"openstack%2Fmonasca-common~734376","triplet_id":"openstack%2Fmonasca-common~master~Ie7239e6651feb358cd0f00fe2709452ebeb3a811","project":"openstack/monasca-common","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"Ie7239e6651feb358cd0f00fe2709452ebeb3a811","subject":"drop mock from lower-constraints","status":"MERGED","created":"2020-06-08 23:24:21.000000000","updated":"2020-06-10 10:55:48.000000000","submitted":"2020-06-10 10:55:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734376-1591786548162-db77c85b","meta_rev_id":"c2f604ba20752149148760650a0680f47c752336","_number":734376,"virtual_id_number":734376,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 10:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-10 10:15:29.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-10 10:15:29.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 10:15:29.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2020-06-10 10:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"952839f163478ee46d058b4c68eba50e8ffdfaec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 23:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a3c2df5159d4034ebb26ee11f4502d640b925b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 00:01:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0908fb0734141d9ae24f83cfd820bd4 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994ccdee78fe45daa36858d2548d3ba0 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f2d92ffad0f4d099962214b96959d13 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe1eb6011f246d792d874ccac89f0f0 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8241e94b03574719a516e8987aa16889 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aa7ed9f48f44f6ba3600d34a2d9836b : SUCCESS in 6m 37s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/0c519fa1f4d44acda7026e60c1ee556b : SUCCESS in 35m 11s\n- monasca-tempest-python3-cassandra https://zuul.opendev.org/t/openstack/build/eca76497a2d74dcdadb5689754abfca2 : FAILURE in 29m 33s (non-voting)\n- build-monasca-common-docker-base-image https://zuul.opendev.org/t/openstack/build/b7a34435e0214477902faea6fe16d969 : SUCCESS in 3m 21s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ed18555eb841415db123d0a826a7230b : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"63378a69faa6c4f9fa4eeb718e36efefa2ae1bd1","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-06-10 10:15:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4246d5b319850c23f61a91c0751e849a4e81f079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:15:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7290d36cb37fee5e38d58395caf69b056130eab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:55:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/52da2514b5df4502a62599271276c392 : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c457affc53e44c439f6bc56e5175c890 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87741387dd4040a683b809fcb8628bb6 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d3226f1afe4ae683d8329be0c391e5 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11e4e7d792b45bc80d51ef4afb36214 : SUCCESS in 4m 27s\n- monasca-tempest-python3-influxdb https://zuul.opendev.org/t/openstack/build/3ef161fd51f5464c86e205f47cd57ab4 : SUCCESS in 35m 10s\n- legacy-monasca-common-maven-build https://zuul.opendev.org/t/openstack/build/ede6d48ed89f4d68ab5e84a81cc335ba : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f4059ad29a9a4d9ccba67d7be675ae64e4284a8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:55:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a7030d1684c3f3ca98007818db62e5543a3381f","revisions":{"3a7030d1684c3f3ca98007818db62e5543a3381f":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:24:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/76/734376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/monasca-common","ref":"refs/changes/76/734376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/734376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/734376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/monasca-common refs/changes/76/734376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/monasca-common refs/changes/76/734376/1"}}},"commit":{"parents":[{"commit":"a1593cd0ff00015b6c06dbf0ae4d246d3650696a","subject":"Stop to use the __future__ module.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/a1593cd0ff00015b6c06dbf0ae4d246d3650696a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:15.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:24:15.000000000","tz":120},"subject":"drop mock from lower-constraints","message":"drop mock from lower-constraints\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we don\u0027t need it\nin lower-constraints.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: Ie7239e6651feb358cd0f00fe2709452ebeb3a811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3a7030d1684c3f3ca98007818db62e5543a3381f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/monasca-common/commit/3a7030d1684c3f3ca98007818db62e5543a3381f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
